)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~654533","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Fstein~I7df8237c0d3c2d690ad6aa68aa96d9b79f422dcb","project":"openstack/openstack-ansible-os_swift","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I7df8237c0d3c2d690ad6aa68aa96d9b79f422dcb","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 15:39:59.000000000","updated":"2019-04-22 20:04:26.000000000","submitted":"2019-04-22 20:04:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654533-1555963466134-a663f06a","meta_rev_id":"24609a65258c2d98fd3ccf4d6f4c4324a38d476b","_number":654533,"virtual_id_number":654533,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-22 20:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-22 16:16:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-22 16:16:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:16:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-22 20:04:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38067ab5ec7dd4a68d5649f49c7f65b1d50c1539","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 15:39:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00d893c615712849b40b7658224b77bdce3cb758","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:16:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"894d65f8aaed909aa87f54beccf2b2b693ffd7c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:34:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/33/654533/1/check/openstack-ansible-linters/f2401c8/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/33/654533/1/check/openstack-tox-docs/ab93ff7/html/ : SUCCESS in 2m 30s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/33/654533/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/64e0e41/ : FAILURE in 28m 41s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/33/654533/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/5e49ec8/ : FAILURE in 21m 36s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/33/654533/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/030db53/ : FAILURE in 21m 25s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/33/654533/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/7bd70cd/ : FAILURE in 22m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b26b4dab532f03f9a665e4fbed676ef47e083276","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:44:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d95a67102b684f4ff90f26dd1e667519d5de210b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:34:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/33/654533/1/check/openstack-ansible-linters/19df8b9/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/33/654533/1/check/openstack-tox-docs/7efcde3/html/ : SUCCESS in 2m 39s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/33/654533/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/11bfcb9/ : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/33/654533/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/d9d2b85/ : SUCCESS in 57m 49s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/33/654533/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/2cd5ea3/ : SUCCESS in 1h 00m 03s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/33/654533/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/b271937/ : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":1},{"id":"19e0a2db7927228ed77b09f0c23ad9f3900d70c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:34:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f5cfb67ece55d86ca939dee05080ed4eef0091a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:04:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/33/654533/1/gate/openstack-ansible-linters/f008081/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/33/654533/1/gate/openstack-tox-docs/296c01a/html/ : SUCCESS in 2m 41s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/33/654533/1/gate/openstack-ansible-deploy-aio_metal_swift-centos-7/1de80e8/ : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/33/654533/1/gate/openstack-ansible-deploy-aio_metal_swift-debian-stable/bb5911e/ : SUCCESS in 54m 06s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/33/654533/1/gate/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/c1ed6a5/ : SUCCESS in 59m 35s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/33/654533/1/gate/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/749248e/ : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":1},{"id":"77f0e9aa643a7183d2943f2fea3c4c1ce0a26f1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:04:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2158b37f08b5bd99d08ee6b0b28100440573f98e","revisions":{"2158b37f08b5bd99d08ee6b0b28100440573f98e":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:39:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/654533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/33/654533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/33/654533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/33/654533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/33/654533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/33/654533/1"}}},"commit":{"parents":[{"commit":"386b089ced4b49aab7e859b0ad5462718b9c1c1c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/386b089ced4b49aab7e859b0ad5462718b9c1c1c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:39:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:39:57.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I7df8237c0d3c2d690ad6aa68aa96d9b79f422dcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2158b37f08b5bd99d08ee6b0b28100440573f98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2158b37f08b5bd99d08ee6b0b28100440573f98e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
