)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~654918","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Fstein~I9f66cd9375942da3921b5e4293d93ce101f6ed66","project":"openstack/openstack-ansible-os_swift","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I9f66cd9375942da3921b5e4293d93ce101f6ed66","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-23 09:24:46.000000000","updated":"2019-04-24 08:43:39.000000000","submitted":"2019-04-24 08:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654918-1556095419214-eaa21e4c","meta_rev_id":"c4d68f6bd26f9e94d5a38b7fca3db5ade0b69fc4","_number":654918,"virtual_id_number":654918,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-24 08:43:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-23 20:22:33.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-23 10:42:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-23 10:42:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 10:42:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-23 20:22:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-24 08:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31b55809e9e15012b320b587972ec8affe8d5c02","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-23 09:24:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e9e7ebca67ea592b1a7ea54b2163024e06b2c15","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-23 10:42:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28b403a7cd106d787e48ad162c603b046e02ac35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:05:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/18/654918/1/check/openstack-ansible-linters/c5fe07d/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/18/654918/1/check/openstack-tox-docs/969970d/html/ : SUCCESS in 2m 37s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/a728c85/ : SUCCESS in 1h 20m 01s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/76b8cd0/ : SUCCESS in 1h 00m 20s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/febb9d6/ : SUCCESS in 1h 00m 02s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/a609681/ : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":1},{"id":"397a63cadcd13bad29cb5f95a2951a81d0c08fc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:05:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"394691190f470f637ce1dc70d46ca253d4ade6c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:11:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/18/654918/1/gate/openstack-ansible-linters/ba2a81f/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/18/654918/1/gate/openstack-tox-docs/84fb7c3/html/ : SUCCESS in 3m 04s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 finger://ze09.openstack.org/c003c12646144a34b7a7fe0fc0eaac33 : RETRY_LIMIT in 32m 41s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable finger://ze12.openstack.org/1ac7354e022241f08690531489f9b733 : RETRY_LIMIT in 42m 27s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic finger://ze03.openstack.org/905233c6083d4f1691f6873a12fd29ae : RETRY_LIMIT in 32m 33s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 finger://ze12.openstack.org/4781204d72524b059fe80ca2d8f61f29 : RETRY_LIMIT in 39m 39s","accounts_in_message":[],"_revision_number":1},{"id":"78d76b3facfdf1363782bfa57681d6b0833e32b2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-23 20:22:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"413807474ff4f3979e59702f359de945ce1aed6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 22:58:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/18/654918/1/check/openstack-ansible-linters/996519d/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/18/654918/1/check/openstack-tox-docs/d391798/html/ : SUCCESS in 4m 09s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/a3e88c2/ : SUCCESS in 1h 29m 04s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/34add73/ : SUCCESS in 1h 03m 16s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/5ce6bb3/ : SUCCESS in 58m 24s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/10793c6/ : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":1},{"id":"92e71be5cec960e80c92c649ef13237c1e86c7aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 22:58:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"266fae2677f330bf9babd26ce2c8b1b156a59214","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 00:09:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/18/654918/1/gate/openstack-ansible-linters/efb5a72/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/18/654918/1/gate/openstack-tox-docs/fe4606c/html/ : SUCCESS in 2m 40s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/gate/openstack-ansible-deploy-aio_metal_swift-centos-7/0e8adef/ : POST_FAILURE in 1h 01m 51s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/18/654918/1/gate/openstack-ansible-deploy-aio_metal_swift-debian-stable/9f57d49/ : SUCCESS in 1h 02m 10s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/18/654918/1/gate/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/4ce9da6/ : SUCCESS in 1h 02m 16s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/gate/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/bb64a7e/ : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"78202d707c65ec1651a4a953543c8681cf82247f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-24 05:53:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"edff7c7e4cd169efa485625fdc46ed7d757cf78f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 07:18:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/18/654918/1/check/openstack-ansible-linters/985ecaa/ : SUCCESS in 9m 04s\n- openstack-tox-docs http://logs.openstack.org/18/654918/1/check/openstack-tox-docs/2996f15/html/ : SUCCESS in 2m 17s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/75928d5/ : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/d2e6f09/ : SUCCESS in 52m 06s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/9b301ee/ : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/087ae3c/ : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"bffb073798465a00f3c430e2a2cdfe5f6c222841","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 07:18:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05e0f12a91f0aa2f755def04f5293b107ee9d2d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 08:43:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8e0cfb14daac2b5b2a89ffdd94c55211a3fa6e14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 08:43:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/18/654918/1/gate/openstack-ansible-linters/0ce95e4/ : SUCCESS in 9m 11s\n- openstack-tox-docs http://logs.openstack.org/18/654918/1/gate/openstack-tox-docs/281f155/html/ : SUCCESS in 2m 35s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/gate/openstack-ansible-deploy-aio_metal_swift-centos-7/ca1f496/ : SUCCESS in 1h 18m 14s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/18/654918/1/gate/openstack-ansible-deploy-aio_metal_swift-debian-stable/9bff753/ : SUCCESS in 59m 21s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/18/654918/1/gate/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/c5bf223/ : SUCCESS in 1h 00m 20s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/18/654918/1/gate/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/228f9d6/ : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09acf2c57d2bb8f875b3fb09c1ff71f7de164f59","revisions":{"09acf2c57d2bb8f875b3fb09c1ff71f7de164f59":{"kind":"REWORK","_number":1,"created":"2019-04-23 09:24:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/654918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/18/654918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/18/654918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/18/654918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/18/654918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/18/654918/1"}}},"commit":{"parents":[{"commit":"2158b37f08b5bd99d08ee6b0b28100440573f98e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2158b37f08b5bd99d08ee6b0b28100440573f98e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:24:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:24:45.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I9f66cd9375942da3921b5e4293d93ce101f6ed66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/09acf2c57d2bb8f875b3fb09c1ff71f7de164f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/09acf2c57d2bb8f875b3fb09c1ff71f7de164f59"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
