)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~657268","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Ib003e3d504a5f9a97ce5274e612f85570681b6c8","project":"openstack/openstack-ansible-os_swift","branch":"master","hashtags":[],"change_id":"Ib003e3d504a5f9a97ce5274e612f85570681b6c8","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"ABANDONED","created":"2019-05-05 13:18:48.000000000","updated":"2020-05-26 07:40:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f74a32a0b95744267cd7d69639fc01eba8401f8d","_number":657268,"virtual_id_number":657268,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-08-15 00:40:10.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":-1,"date":"2019-05-06 12:54:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 12:54:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 00:40:10.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"ca338280db80d8f32c07e31061f7d381fb7a2761","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-05 13:18:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5da317ac725e71dec3b38ef2f91ab16c68e59c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 14:49:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/68/657268/1/check/openstack-ansible-linters/32fbd2d/ : SUCCESS in 6m 28s\n- openstack-tox-docs http://logs.openstack.org/68/657268/1/check/openstack-tox-docs/4b0474a/html/ : SUCCESS in 2m 50s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/68/657268/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/e4bc862/ : SUCCESS in 1h 23m 14s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/68/657268/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/265fa6a/ : SUCCESS in 59m 13s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/68/657268/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/0448e16/ : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/68/657268/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/ae0a6c9/ : FAILURE in 37m 17s","accounts_in_message":[],"_revision_number":1},{"id":"240bc9593d96fad1e2483a8ad0ead45a53e897f3","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 11:30:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"27f3332ecc717ed37f3813a3a8209936a2bee65f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:54:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/68/657268/2/check/openstack-ansible-linters/5cc1d1c/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/68/657268/2/check/openstack-tox-docs/4a446fa/html/ : SUCCESS in 2m 53s\n- build-openstack-releasenotes http://logs.openstack.org/68/657268/2/check/build-openstack-releasenotes/ea8c5c9/html/ : SUCCESS in 2m 41s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/68/657268/2/check/openstack-ansible-deploy-aio_metal_swift-centos-7/030091e/ : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/68/657268/2/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/2040b51/ : SUCCESS in 1h 00m 35s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/68/657268/2/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/d32639a/ : SUCCESS in 58m 54s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/68/657268/2/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/4c6e0e3/ : FAILURE in 35m 56s","accounts_in_message":[],"_revision_number":2},{"id":"cd58abac1b982fc375c3af30ffb10bb7627fbc7f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-15 00:40:10.000000000","message":"Patch Set 2:\n\ncan you please rebase this? :)","accounts_in_message":[],"_revision_number":2},{"id":"f74a32a0b95744267cd7d69639fc01eba8401f8d","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-05-26 07:40:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3e7fc3ebbf255901cbb3572809106817dbc7e656","revisions":{"bfb39f79a227634658f030e7f3c0f50859577d9c":{"kind":"REWORK","_number":1,"created":"2019-05-05 13:18:48.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/68/657268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/68/657268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/68/657268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/68/657268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/68/657268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/68/657268/1"}}},"commit":{"parents":[{"commit":"bae76c0d14f39cecfcf85910731abddf9449f13c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/bae76c0d14f39cecfcf85910731abddf9449f13c"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:18:33.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:18:46.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ib003e3d504a5f9a97ce5274e612f85570681b6c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/bfb39f79a227634658f030e7f3c0f50859577d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/bfb39f79a227634658f030e7f3c0f50859577d9c"}]},"branch":"refs/heads/master"},"3e7fc3ebbf255901cbb3572809106817dbc7e656":{"kind":"REWORK","_number":2,"created":"2019-05-06 11:30:36.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/68/657268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/68/657268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/68/657268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/68/657268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/68/657268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/68/657268/2"}}},"commit":{"parents":[{"commit":"bae76c0d14f39cecfcf85910731abddf9449f13c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/bae76c0d14f39cecfcf85910731abddf9449f13c"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 13:18:33.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 11:30:20.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ib003e3d504a5f9a97ce5274e612f85570681b6c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3e7fc3ebbf255901cbb3572809106817dbc7e656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3e7fc3ebbf255901cbb3572809106817dbc7e656"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
