)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~658614","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Fstein~I639c70430a250831f38bd0768e572300a6a6fa63","project":"openstack/openstack-ansible-os_swift","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I639c70430a250831f38bd0768e572300a6a6fa63","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:59:42.000000000","updated":"2019-07-15 22:15:52.000000000","submitted":"2019-07-15 22:15:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658614-1563228952405-fd40a304","meta_rev_id":"756e0970a33d2002c1cd1e365591584d1a06743a","_number":658614,"virtual_id_number":658614,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"date":"2019-06-27 10:22:55.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-15 22:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-20 20:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-15 20:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-07-15 20:32:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 14:40:05.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-20 20:11:28.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-27 10:22:55.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-15 20:32:55.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-15 22:15:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7857de0b1fe20199ea6118a15a59ddf5177a79bc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:59:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6bb41a53420baafc74a62f3e37b270ab8e12a1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:36:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/658614/1/check/openstack-ansible-linters/6939718/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/14/658614/1/check/openstack-tox-docs/259a0a3/html/ : SUCCESS in 3m 35s\n- build-openstack-releasenotes http://logs.openstack.org/14/658614/1/check/build-openstack-releasenotes/5e98ef8/html/ : SUCCESS in 3m 51s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/3531bb5/ : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/5f6a1d6/ : SUCCESS in 1h 02m 21s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/5e92681/ : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/60c3707/ : FAILURE in 37m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2e98ad408a62012d2147749625f3494f9a6c4f53","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-04 14:40:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb21b11ed77403d72b5e554f8995c846f8951521","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 16:12:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/658614/1/check/openstack-ansible-linters/d7f9cd0/ : SUCCESS in 6m 11s\n- openstack-tox-docs http://logs.openstack.org/14/658614/1/check/openstack-tox-docs/4bf2988/html/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/14/658614/1/check/build-openstack-releasenotes/0f613c5/html/ : SUCCESS in 2m 44s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/c4b3972/ : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/294c4c0/ : SUCCESS in 1h 04m 02s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/6f23b1c/ : SUCCESS in 59m 27s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/34b21e9/ : FAILURE in 41m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cc5519f8915b9cebd15562744d36541d014e3918","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-06 06:52:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f79ac094d13cd95c6e3297b1ebd19e833feb713e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 08:43:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/658614/1/check/openstack-ansible-linters/2c10fb3/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/14/658614/1/check/openstack-tox-docs/9ff4644/html/ : SUCCESS in 2m 33s\n- build-openstack-releasenotes http://logs.openstack.org/14/658614/1/check/build-openstack-releasenotes/f0df3c0/html/ : SUCCESS in 2m 16s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/a0b44b1/ : SUCCESS in 1h 32m 32s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/5a2be47/ : SUCCESS in 1h 04m 43s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/9eebee3/ : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/14/658614/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/80eb402/ : FAILURE in 38m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4916236e378d5d32f45d00bccbc328d4bd45dd05","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-20 20:06:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e16145e3155ba402335434536ea5e819db1b596f","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-20 20:11:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a829b9d12d2965d1d2cce109010b92f6aeaa1884","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:30:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/658614/2/check/openstack-ansible-linters/f9761f3/ : SUCCESS in 14m 43s\n- openstack-tox-docs http://logs.openstack.org/14/658614/2/check/openstack-tox-docs/2f38e74/html/ : SUCCESS in 6m 10s\n- build-openstack-releasenotes http://logs.openstack.org/14/658614/2/check/build-openstack-releasenotes/a5ec1de/html/ : SUCCESS in 3m 28s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_metal_swift-centos-7/3d75f6f/ : SUCCESS in 1h 21m 38s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/8c2e20c/ : SUCCESS in 1h 00m 46s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/67b8764/ : SUCCESS in 1h 47m 03s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/f2e7e2b/ : FAILURE in 37m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0cb688629620e46a772e21f9585692af36ac346d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-24 09:12:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8d6a6e6dbd4956512db31db34142072a5c42413b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 11:48:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/14/658614/2/check/openstack-ansible-linters/49f2e2b/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/14/658614/2/check/openstack-tox-docs/078a9ed/html/ : SUCCESS in 2m 25s\n- build-openstack-releasenotes http://logs.openstack.org/14/658614/2/check/build-openstack-releasenotes/bfde27b/html/ : SUCCESS in 3m 27s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_metal_swift-centos-7/5f4adac/ : SUCCESS in 1h 29m 54s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/3a2b6ad/ : SUCCESS in 1h 05m 08s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/8fe3a41/ : SUCCESS in 1h 01m 31s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/4613772/ : FAILURE in 47m 39s","accounts_in_message":[],"_revision_number":2},{"id":"068d03f493498df1ea44f9c2cb2c92c33c4145f5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-27 10:22:55.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"e42dfd92151654482771a74e5423bcf810bf528c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:19:49.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\n\n- openstack-ansible-deploy-aio_metal_swift-opensuse-150 http://logs.openstack.org/14/658614/2/experimental/openstack-ansible-deploy-aio_metal_swift-opensuse-150/38eee2e/ : FAILURE in 18m 59s\n- openstack-ansible-deploy-aio_distro_metal_swift-opensuse-150 http://logs.openstack.org/14/658614/2/experimental/openstack-ansible-deploy-aio_distro_metal_swift-opensuse-150/338e67c/ : FAILURE in 15m 57s\n- openstack-ansible-deploy-aio_distro_metal_swift-ubuntu-bionic http://logs.openstack.org/14/658614/2/experimental/openstack-ansible-deploy-aio_distro_metal_swift-ubuntu-bionic/11da2cc/ : SUCCESS in 50m 04s","accounts_in_message":[],"_revision_number":2},{"id":"646b1c515bdf46a85ef5d3ffc2310580a3770579","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-15 19:01:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"45422f67cafd30d83467ce8e2f1fc27b7bdbf49a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 20:31:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/14/658614/2/check/openstack-ansible-linters/fee46cc/ : SUCCESS in 9m 00s\n- openstack-tox-docs http://logs.openstack.org/14/658614/2/check/openstack-tox-docs/6b5f8d9/html/ : SUCCESS in 3m 30s\n- build-openstack-releasenotes http://logs.openstack.org/14/658614/2/check/build-openstack-releasenotes/0332936/html/ : SUCCESS in 4m 34s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_metal_swift-centos-7/535c0e8/ : SUCCESS in 1h 25m 55s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/0a46c9c/ : SUCCESS in 1h 11m 31s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/ef06489/ : SUCCESS in 1h 09m 01s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/14/658614/2/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/9cf5d11/ : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":2},{"id":"7d1d08376b01e9ef1ecc662523dc6dafa6a2be38","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-15 20:32:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f59a8d9ca1912f5f0a6e243f7181030cda399e1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 20:33:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"379e8a6405a165c8f5b7a46388a025b1f8e6fd12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 22:15:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/14/658614/2/gate/openstack-ansible-linters/9e0b238/ : SUCCESS in 6m 47s\n- openstack-tox-docs http://logs.openstack.org/14/658614/2/gate/openstack-tox-docs/6e136af/html/ : SUCCESS in 2m 56s\n- build-openstack-releasenotes http://logs.openstack.org/14/658614/2/gate/build-openstack-releasenotes/ad8bc5c/html/ : SUCCESS in 3m 45s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/14/658614/2/gate/openstack-ansible-deploy-aio_metal_swift-centos-7/80ef150/ : SUCCESS in 1h 37m 16s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/14/658614/2/gate/openstack-ansible-deploy-aio_metal_swift-debian-stable/223c48d/ : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/14/658614/2/gate/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/c32d0a5/ : SUCCESS in 1h 04m 53s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/14/658614/2/gate/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/b33c40a/ : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b20fda06535a6a08196af5973334eec46869adfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 22:15:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0fc8081b9ca69d4a2581e5c3b4eadf129974cc81","revisions":{"cf7bd856ed51498f7ae00a702d6354cbe883cc2d":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:59:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/658614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/14/658614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/14/658614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/14/658614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/14/658614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/14/658614/1"}}},"commit":{"parents":[{"commit":"09acf2c57d2bb8f875b3fb09c1ff71f7de164f59","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/09acf2c57d2bb8f875b3fb09c1ff71f7de164f59"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:41.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I639c70430a250831f38bd0768e572300a6a6fa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/cf7bd856ed51498f7ae00a702d6354cbe883cc2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/cf7bd856ed51498f7ae00a702d6354cbe883cc2d"}]},"branch":"refs/heads/stable/stein"},"0fc8081b9ca69d4a2581e5c3b4eadf129974cc81":{"kind":"REWORK","_number":2,"created":"2019-06-20 20:06:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/658614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/14/658614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/14/658614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/14/658614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/14/658614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/14/658614/2"}}},"commit":{"parents":[{"commit":"09acf2c57d2bb8f875b3fb09c1ff71f7de164f59","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/09acf2c57d2bb8f875b3fb09c1ff71f7de164f59"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:06:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:06:44.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I639c70430a250831f38bd0768e572300a6a6fa63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/0fc8081b9ca69d4a2581e5c3b4eadf129974cc81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/0fc8081b9ca69d4a2581e5c3b4eadf129974cc81"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
