)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~663569","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Ie7ae55e68db100e712e6d862aefec2220dc0a83b","project":"openstack/openstack-ansible-os_swift","branch":"master","hashtags":[],"change_id":"Ie7ae55e68db100e712e6d862aefec2220dc0a83b","subject":"Cap sphinx for py2 to match global requirements","status":"MERGED","created":"2019-06-06 09:49:09.000000000","updated":"2019-07-10 18:52:25.000000000","submitted":"2019-07-10 18:52:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663569-1562784745229-0a1ee91c","meta_rev_id":"8532e5ce2b7e52829d943d322ae7215768287a9f","_number":663569,"virtual_id_number":663569,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-10 18:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-10 18:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-10 18:38:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 18:38:33.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-10 18:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5ed57c7f107c812a4d64729bb9dbb6c60522826","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-06-06 09:49:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfe2e59b8407abf466b08bf27976267f9466f259","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:00:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/69/663569/1/check/requirements-check/e12aa91/ : SUCCESS in 2m 06s\n- openstack-ansible-linters http://logs.openstack.org/69/663569/1/check/openstack-ansible-linters/6f660c6/ : SUCCESS in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/69/663569/1/check/openstack-tox-docs/c15488d/html/ : SUCCESS in 2m 43s\n- build-openstack-releasenotes http://logs.openstack.org/69/663569/1/check/build-openstack-releasenotes/8911a19/html/ : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d0c4a2eb21214f0fa84e1b518a666469f2f2f059","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 18:38:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a604134c2a9d483bd4f8dd4f1eebe9ce8f18ae00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:38:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a610d10b57de22079d747e5e880f3b75a460e91f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:52:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8532e5ce2b7e52829d943d322ae7215768287a9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:52:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/69/663569/1/gate/requirements-check/657ee5f/ : SUCCESS in 2m 23s\n- openstack-ansible-linters http://logs.openstack.org/69/663569/1/gate/openstack-ansible-linters/31c7abe/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/69/663569/1/gate/openstack-tox-docs/9d407b5/html/ : SUCCESS in 4m 25s\n- build-openstack-releasenotes http://logs.openstack.org/69/663569/1/gate/build-openstack-releasenotes/b634001/html/ : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27e48fc1aee667eced1c77d62519dfa05c8f3bbb","revisions":{"27e48fc1aee667eced1c77d62519dfa05c8f3bbb":{"kind":"REWORK","_number":1,"created":"2019-06-06 09:49:09.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/69/663569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/69/663569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/69/663569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/69/663569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/69/663569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/69/663569/1"}}},"commit":{"parents":[{"commit":"d690240354758b6e861c93f67924bbdcc2ced376","subject":"Clarify supported SUSE versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d690240354758b6e861c93f67924bbdcc2ced376"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-06 09:48:44.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-06 09:49:04.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: Ie7ae55e68db100e712e6d862aefec2220dc0a83b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/27e48fc1aee667eced1c77d62519dfa05c8f3bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/27e48fc1aee667eced1c77d62519dfa05c8f3bbb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
