)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~664104","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Fqueens~Icbc172f7c4f2494bd6feefb97b8b81434b1b46c7","project":"openstack/openstack-ansible-os_swift","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Icbc172f7c4f2494bd6feefb97b8b81434b1b46c7","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-07 23:28:38.000000000","updated":"2019-06-11 21:34:25.000000000","submitted":"2019-06-11 21:34:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664104-1560288865830-e5d355c6","meta_rev_id":"3a66618a666374245a268a141018a2a1ed160c09","_number":664104,"virtual_id_number":664104,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-11 21:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 20:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 20:56:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 20:56:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 21:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb1c1825f48d5479dd1200dde3fdce22e6abe044","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:28:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f83b13163dfcca90fabd938cd0d00670c2ac4b4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:48:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/04/664104/1/check/openstack-ansible-linters/00f2b9c/ : SUCCESS in 5m 10s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/04/664104/1/check/openstack-ansible-functional-centos-7/0a3ff18/ : SUCCESS in 52m 46s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/04/664104/1/check/openstack-ansible-functional-opensuse-423/9633cd3/ : FAILURE in 9m 29s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/04/664104/1/check/openstack-ansible-functional-ubuntu-xenial/a4906d4/ : SUCCESS in 34m 15s\n- openstack-ansible-swift3-ubuntu-xenial http://logs.openstack.org/04/664104/1/check/openstack-ansible-swift3-ubuntu-xenial/7d19c21/ : SUCCESS in 30m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c38227eed0caea0b600acfe9a99d44626ad60ea5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 19:31:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"560f6a28a41421c8643cb31a6be0f50a2f3029e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:33:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/04/664104/1/check/openstack-ansible-linters/b2957ec/ : SUCCESS in 6m 18s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/04/664104/1/check/openstack-ansible-functional-centos-7/a47a366/ : SUCCESS in 55m 46s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/04/664104/1/check/openstack-ansible-functional-ubuntu-xenial/d1898b2/ : SUCCESS in 31m 27s\n- openstack-ansible-swift3-ubuntu-xenial http://logs.openstack.org/04/664104/1/check/openstack-ansible-swift3-ubuntu-xenial/929cec7/ : SUCCESS in 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b900db94ad5fcb0ce73f3b8aeea3f8bb748d2a43","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 20:56:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b7f93e8439462dfc6782f58f3d12c4518e3baba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:56:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e62cfa654ebb1257c42fcb809098d80edfdb40a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:34:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/04/664104/1/gate/openstack-ansible-linters/f5bd95d/ : SUCCESS in 4m 56s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/04/664104/1/gate/openstack-ansible-functional-ubuntu-xenial/9058adf/ : SUCCESS in 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"13c819525fb98b61555d62adbf5e37bc79fa6974","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:34:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98624f55d28609576fdcbcec3a0a13849676542d","revisions":{"98624f55d28609576fdcbcec3a0a13849676542d":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:28:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/664104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/04/664104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/04/664104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/04/664104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/04/664104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/04/664104/1"}}},"commit":{"parents":[{"commit":"c8920ec66d3c9f8220d695335c0f687d30a9defa","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c8920ec66d3c9f8220d695335c0f687d30a9defa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:28:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:28:33.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Icbc172f7c4f2494bd6feefb97b8b81434b1b46c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/98624f55d28609576fdcbcec3a0a13849676542d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/98624f55d28609576fdcbcec3a0a13849676542d"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
