)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~664344","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Frocky~I5961025aca9ebedf0b68e289ab13632eb974c997","project":"openstack/openstack-ansible-os_swift","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I5961025aca9ebedf0b68e289ab13632eb974c997","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:45:13.000000000","updated":"2019-06-10 18:57:29.000000000","submitted":"2019-06-10 18:57:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664344-1560193049103-8840775f","meta_rev_id":"2c927002e17d71167a02b249875f9baef07d53c8","_number":664344,"virtual_id_number":664344,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2019-06-10 18:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2019-06-10 17:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2019-06-10 17:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:35:06.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-06-10 18:57:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16b16f0d33d596111d24d2bc38f6bd366f6dc999","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:45:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"437a95fab1f5c20a659b88b04cba6654bc2a459e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:19:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/44/664344/1/check/openstack-ansible-linters/20b26fe/ : SUCCESS in 4m 56s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/664344/1/check/openstack-ansible-functional-centos-7/8ad4356/ : SUCCESS in 1h 02m 40s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/664344/1/check/openstack-ansible-functional-opensuse-423/531442c/ : FAILURE in 8m 37s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/44/664344/1/check/openstack-ansible-functional-opensuse-150/7cc61a7/ : FAILURE in 13m 57s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/664344/1/check/openstack-ansible-functional-ubuntu-bionic/e29593a/ : SUCCESS in 54m 50s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/44/664344/1/check/openstack-ansible-functional-ubuntu-xenial/f0898a0/ : SUCCESS in 1h 00m 39s\n- openstack-ansible-swift3-ubuntu-xenial http://logs.openstack.org/44/664344/1/check/openstack-ansible-swift3-ubuntu-xenial/cd9b56e/ : FAILURE in 1h 07m 48s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/44/664344/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/8056817/ : SUCCESS in 33m 28s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/44/664344/1/check/openstack-ansible-functional-distro_install-centos-7/fe957e2/ : FAILURE in 30m 58s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/664344/1/check/openstack-ansible-functional-distro_install-opensuse-423/869c0d3/ : FAILURE in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4383a41978745ad0a582b514e6031d05af9a326","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-10 17:34:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c19ec3cef3642050924c5bd4d63b1aaa3b98477e","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-10 17:35:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e96561012041a0b8d4a36bb07e448b1dbad98100","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:35:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fba13941ace01941164cb7e3f33a155f68a4a962","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:57:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/44/664344/1/gate/openstack-ansible-linters/9809d3e/ : SUCCESS in 6m 36s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/44/664344/1/gate/openstack-ansible-functional-centos-7/1cef1ef/ : SUCCESS in 1h 01m 27s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/44/664344/1/gate/openstack-ansible-functional-opensuse-423/7e2c068/ : FAILURE in 8m 29s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/44/664344/1/gate/openstack-ansible-functional-ubuntu-bionic/ae36d2f/ : SUCCESS in 51m 05s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/44/664344/1/gate/openstack-ansible-functional-ubuntu-xenial/6be4032/ : SUCCESS in 1h 08m 54s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/44/664344/1/gate/openstack-ansible-functional-distro_install-ubuntu-xenial/f5bd299/ : SUCCESS in 38m 22s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/44/664344/1/gate/openstack-ansible-functional-distro_install-opensuse-423/fe1e564/ : FAILURE in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"835a76c73d09450e959ea5b71b676b15d4818332","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:57:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c269b29e690b8f9bedc8e25ef0bcf65d585c3ba9","revisions":{"c269b29e690b8f9bedc8e25ef0bcf65d585c3ba9":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:45:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/664344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/44/664344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/44/664344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/44/664344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/44/664344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/44/664344/1"}}},"commit":{"parents":[{"commit":"9e4283e8e53c351f3d1ee8cd28f369969d5c8d22","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/9e4283e8e53c351f3d1ee8cd28f369969d5c8d22"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:12.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I5961025aca9ebedf0b68e289ab13632eb974c997\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c269b29e690b8f9bedc8e25ef0bcf65d585c3ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c269b29e690b8f9bedc8e25ef0bcf65d585c3ba9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
