)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~664644","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I1e7a8eba6500c819d418ee565a5f03cc16dfe165","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"distro-install","hashtags":[],"change_id":"I1e7a8eba6500c819d418ee565a5f03cc16dfe165","subject":"Use groups to determine distro packages to install","status":"MERGED","created":"2019-06-11 17:39:27.000000000","updated":"2019-06-11 23:00:31.000000000","submitted":"2019-06-11 22:58:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664644-1560293891497-91b8215b","meta_rev_id":"136b5f2af26bcba39a072498ca18184d5c381682","_number":664644,"virtual_id_number":664644,"owner":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-11 22:58:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-11 20:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 21:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 21:12:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 20:38:04.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-11 21:12:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 22:58:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68e0457e256590c31524fcd528a65b1b59398649","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-11 17:39:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6fb469cf5b01d5ecfe6148e3703446d972b9977","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 19:26:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/44/664644/1/check/openstack-ansible-linters/6f84d71/ : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/44/664644/1/check/openstack-ansible-deploy-aio_metal-debian-stable/4f98b50/ : SUCCESS in 1h 08m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/44/664644/1/check/openstack-ansible-deploy-aio_metal-centos-7/40b77d2/ : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/44/664644/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/c5a980d/ : SUCCESS in 1h 20m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/44/664644/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/7a61dbf/ : SUCCESS in 1h 00m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/44/664644/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/04f4250/ : SUCCESS in 1h 13m 05s\n- openstack-tox-docs http://logs.openstack.org/44/664644/1/check/openstack-tox-docs/16b3f8e/html/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"861052310ed2e0bd344d91aa48b147f33648c2f1","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-11 20:38:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4e3b985c1da7c04eaf11a84f7ce0dc5b0e1906d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 21:12:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"007e8787b05c798d28748baae0332c677a3e12e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:13:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed1121ffc9541cbb53cb304fd596a113ff8962e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:58:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"13d2a5675f76fff1851ca606bc2aa385b8d354a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:58:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/44/664644/1/gate/openstack-ansible-linters/c46af88/ : SUCCESS in 5m 24s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/44/664644/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/614476e/ : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/44/664644/1/gate/openstack-ansible-deploy-aio_metal-centos-7/98838e3/ : SUCCESS in 1h 29m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/44/664644/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/38b5a44/ : SUCCESS in 1h 18m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/44/664644/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/51bf0c5/ : SUCCESS in 1h 13m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/44/664644/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/065c1a4/ : SUCCESS in 1h 11m 28s\n- openstack-tox-docs http://logs.openstack.org/44/664644/1/gate/openstack-tox-docs/c6dfb3e/html/ : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2e1428a8419a05d553e247540e5655d5aab0a67a","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-11 23:00:31.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 2c2a36be3e157e53e2c4828e4f1bbe7a3fc23fea","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d9567ebadd65279353c90cc9dc1ea3f1f023bc6","revisions":{"3d9567ebadd65279353c90cc9dc1ea3f1f023bc6":{"kind":"REWORK","_number":1,"created":"2019-06-11 17:39:27.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/44/664644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/44/664644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/44/664644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/44/664644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/44/664644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/44/664644/1"}}},"commit":{"parents":[{"commit":"d690240354758b6e861c93f67924bbdcc2ced376","subject":"Clarify supported SUSE versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d690240354758b6e861c93f67924bbdcc2ced376"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-11 17:34:21.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-11 17:38:54.000000000","tz":-420},"subject":"Use groups to determine distro packages to install","message":"Use groups to determine distro packages to install\n\nUse the groups defined in swift_services to determine which distro\npackages to install onto hosts instead of installing all distro service\npackages on all swift hosts.\n\nThe \u0027swift-plugin-s3\u0027 package has also been removed from debian\u0027s\npackage list. s3api should be configured through swift middleware instead.\n\nChange-Id: I1e7a8eba6500c819d418ee565a5f03cc16dfe165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3d9567ebadd65279353c90cc9dc1ea3f1f023bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3d9567ebadd65279353c90cc9dc1ea3f1f023bc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
