)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~664732","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I9354f0cc362580874b7fcb873c744e83e05aa7d2","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I9354f0cc362580874b7fcb873c744e83e05aa7d2","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-11 22:19:26.000000000","updated":"2019-06-12 07:59:03.000000000","submitted":"2019-06-12 07:59:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664732-1560326343646-7fce86ad","meta_rev_id":"97ecec4f6d39676ec248d69ccc3d7c199365787d","_number":664732,"virtual_id_number":664732,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-12 07:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-12 06:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-12 06:14:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 06:14:06.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-12 07:59:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"269da8784522ad1342ed6c6e8167d2a2509389bf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-11 22:19:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4254627705125c76495de7d4ce1773ecc2658849","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:03:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/32/664732/1/check/openstack-ansible-linters/4dba1eb/ : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/32/664732/1/check/openstack-ansible-deploy-aio_metal-debian-stable/5d08580/ : SUCCESS in 1h 08m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/32/664732/1/check/openstack-ansible-deploy-aio_metal-centos-7/f13d20b/ : SUCCESS in 1h 34m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/32/664732/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/c39fcb0/ : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/32/664732/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/72b1831/ : SUCCESS in 1h 03m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/32/664732/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8936dde/ : SUCCESS in 1h 07m 58s\n- openstack-tox-docs http://logs.openstack.org/32/664732/1/check/openstack-tox-docs/b4d35be/html/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"21032a5fd1f8d9e80f0bb8b50ae7b1460944cbf4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-12 06:14:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fcfeb6993889e914bf17157b6eff42b347c87d49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 06:14:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a2bf04a423894f97d14a2876e74f71dbed22c1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:59:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f45b94819be462ad889413d1cfc74d6ba44ca860","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:59:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/32/664732/1/gate/openstack-ansible-linters/73e315a/ : SUCCESS in 6m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/32/664732/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/9e0cf7b/ : SUCCESS in 1h 00m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/32/664732/1/gate/openstack-ansible-deploy-aio_metal-centos-7/eac6533/ : SUCCESS in 1h 24m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/32/664732/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/d5b599d/ : SUCCESS in 1h 12m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/32/664732/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5fb35fd/ : SUCCESS in 1h 06m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/32/664732/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/8c76aa7/ : SUCCESS in 1h 15m 27s\n- openstack-tox-docs http://logs.openstack.org/32/664732/1/gate/openstack-tox-docs/5d19f13/html/ : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08f9e7823de26006db126279b6bac3974df064c2","revisions":{"08f9e7823de26006db126279b6bac3974df064c2":{"kind":"REWORK","_number":1,"created":"2019-06-11 22:19:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/664732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/32/664732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/32/664732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/32/664732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/32/664732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/32/664732/1"}}},"commit":{"parents":[{"commit":"d690240354758b6e861c93f67924bbdcc2ced376","subject":"Clarify supported SUSE versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d690240354758b6e861c93f67924bbdcc2ced376"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:19:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:19:24.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I9354f0cc362580874b7fcb873c744e83e05aa7d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/08f9e7823de26006db126279b6bac3974df064c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/08f9e7823de26006db126279b6bac3974df064c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
