)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~664740","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Fstein~I1e7a8eba6500c819d418ee565a5f03cc16dfe165","project":"openstack/openstack-ansible-os_swift","branch":"stable/stein","topic":"distro-install-stable/stein","hashtags":[],"change_id":"I1e7a8eba6500c819d418ee565a5f03cc16dfe165","subject":"Use groups to determine distro packages to install","status":"MERGED","created":"2019-06-11 23:00:31.000000000","updated":"2019-08-30 21:50:56.000000000","submitted":"2019-08-30 21:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664740-1567201754067-610ae198","meta_rev_id":"976b59e5a8198e4b2c77c8ca7267333266d1d36d","_number":664740,"virtual_id_number":664740,"owner":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-08-30 21:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-08-30 20:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-27 15:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2019-08-30 20:06:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 15:22:13.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-30 20:06:00.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-08-30 21:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b8ceb596447051989e9e719fcaa482e97e4d0e4","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-11 23:00:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"83580214a5a5074f9dd8b5e3eebce3d6a2a2a9c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:25:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/40/664740/1/check/openstack-ansible-linters/0cc4fb2/ : SUCCESS in 5m 28s\n- openstack-tox-docs http://logs.openstack.org/40/664740/1/check/openstack-tox-docs/abf2cd0/html/ : SUCCESS in 3m 00s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/40/664740/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/aae93cd/ : SUCCESS in 1h 20m 07s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/40/664740/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/1f6edae/ : SUCCESS in 1h 06m 39s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/40/664740/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/76c4e7f/ : SUCCESS in 56m 30s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/40/664740/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/03e43fe/ : FAILURE in 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8fa78659f86fb307e4c78620652c1c0b74e94dee","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-13 17:40:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"36875dcc70ad5c72b64bb749c5f38588d7cf01a1","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-13 18:43:38.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d9173be3fbdfec1d927dabb70bf6485f47b89bfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 20:34:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/40/664740/3/check/openstack-ansible-linters/b526797/ : SUCCESS in 5m 59s\n- openstack-tox-docs http://logs.openstack.org/40/664740/3/check/openstack-tox-docs/d103ede/html/ : SUCCESS in 3m 01s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/40/664740/3/check/openstack-ansible-deploy-aio_metal_swift-centos-7/a34f727/ : SUCCESS in 1h 24m 28s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/40/664740/3/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/aaab8e9/ : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/40/664740/3/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/85a1ada/ : SUCCESS in 58m 09s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/40/664740/3/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/a2d5213/ : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":3},{"id":"b12a7678d81a2b645f8e72608f3ba3456b8ac194","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-27 15:22:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"321e8c71d8d7b18f67cb91747be11b902584856e","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-08-30 20:06:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"df6e32292c6ca06bc7a216b9f1431f8478424ae0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:06:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0cef42c7c702d55b80f0dcdb355f7ffa9f26ff82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 21:49:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e9d17380cc064c9c888da38d1a968b52 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249508c736d64661ae5a14a655286420 : SUCCESS in 3m 57s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 https://zuul.opendev.org/t/openstack/build/8eea9a4989c343e6a21da426d9f746fc : SUCCESS in 1h 29m 36s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable https://zuul.opendev.org/t/openstack/build/880dd0ed8dea4aae91121d503823865b : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dda5b8ebb75c446d8abdf93a23bac8d8 : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 https://zuul.opendev.org/t/openstack/build/8a6fd1fd59664bbb8a12e3d02d405822 : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7d819c8ef37b64e43b8ed422eb16c865689a2204","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 21:49:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7409460c8864b7ba824ca0f6c05483a6544b43e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 21:50:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2928a983b7e47d6907a4e965f9415ae : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c3d1a5bed21f56f031ef6fcfe885b3a5dec46929","revisions":{"2c2a36be3e157e53e2c4828e4f1bbe7a3fc23fea":{"kind":"REWORK","_number":1,"created":"2019-06-11 23:00:31.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/40/664740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/40/664740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/1"}}},"commit":{"parents":[{"commit":"09acf2c57d2bb8f875b3fb09c1ff71f7de164f59","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/09acf2c57d2bb8f875b3fb09c1ff71f7de164f59"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-11 17:34:21.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-11 23:00:31.000000000","tz":0},"subject":"Use groups to determine distro packages to install","message":"Use groups to determine distro packages to install\n\nUse the groups defined in swift_services to determine which distro\npackages to install onto hosts instead of installing all distro service\npackages on all swift hosts.\n\nThe \u0027swift-plugin-s3\u0027 package has also been removed from debian\u0027s\npackage list. s3api should be configured through swift middleware instead.\n\nChange-Id: I1e7a8eba6500c819d418ee565a5f03cc16dfe165\n(cherry picked from commit 3d9567ebadd65279353c90cc9dc1ea3f1f023bc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2c2a36be3e157e53e2c4828e4f1bbe7a3fc23fea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2c2a36be3e157e53e2c4828e4f1bbe7a3fc23fea"}]},"branch":"refs/heads/stable/stein"},"43a43876e54f126b00e44a72ab452e5286979197":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-13 17:40:45.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/40/664740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/40/664740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/2"}}},"commit":{"parents":[{"commit":"09acf2c57d2bb8f875b3fb09c1ff71f7de164f59","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/09acf2c57d2bb8f875b3fb09c1ff71f7de164f59"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-11 17:34:21.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-13 17:40:43.000000000","tz":0},"subject":"Use groups to determine distro packages to install","message":"Use groups to determine distro packages to install\n\nUse the groups defined in swift_services to determine which distro\npackages to install onto hosts instead of installing all distro service\npackages on all swift hosts.\n\nThe \u0027swift-plugin-s3\u0027 package has also been removed from debian\u0027s\npackage list. s3api should be configured through swift middleware instead.\n\nDepends-On: https://review.opendev.org/#/c/665189\n\nChange-Id: I1e7a8eba6500c819d418ee565a5f03cc16dfe165\n(cherry picked from commit 3d9567ebadd65279353c90cc9dc1ea3f1f023bc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/43a43876e54f126b00e44a72ab452e5286979197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/43a43876e54f126b00e44a72ab452e5286979197"}]},"branch":"refs/heads/stable/stein"},"c3d1a5bed21f56f031ef6fcfe885b3a5dec46929":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-13 18:43:38.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/40/664740/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/40/664740/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/40/664740/3"}}},"commit":{"parents":[{"commit":"09acf2c57d2bb8f875b3fb09c1ff71f7de164f59","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/09acf2c57d2bb8f875b3fb09c1ff71f7de164f59"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-11 17:34:21.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-13 18:43:31.000000000","tz":0},"subject":"Use groups to determine distro packages to install","message":"Use groups to determine distro packages to install\n\nUse the groups defined in swift_services to determine which distro\npackages to install onto hosts instead of installing all distro service\npackages on all swift hosts.\n\nThe \u0027swift-plugin-s3\u0027 package has also been removed from debian\u0027s\npackage list. s3api should be configured through swift middleware instead.\n\nDepends-On: https://review.opendev.org/#/c/665209\n\nChange-Id: I1e7a8eba6500c819d418ee565a5f03cc16dfe165\n(cherry picked from commit 3d9567ebadd65279353c90cc9dc1ea3f1f023bc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c3d1a5bed21f56f031ef6fcfe885b3a5dec46929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c3d1a5bed21f56f031ef6fcfe885b3a5dec46929"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
