)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~668287","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Iff4011c5a6531fa3f74acfeb6fa980c493246a63","project":"openstack/openstack-ansible-os_swift","branch":"master","hashtags":[],"change_id":"Iff4011c5a6531fa3f74acfeb6fa980c493246a63","subject":"Fix distro package lists for openSUSE 15","status":"MERGED","created":"2019-06-29 07:40:59.000000000","updated":"2019-07-02 14:33:12.000000000","submitted":"2019-07-02 14:33:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668287-1562077992183-2ec5262f","meta_rev_id":"5ac310cc6358e5e95e01810624ebcf3fe6fc7a0a","_number":668287,"virtual_id_number":668287,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-02 14:33:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":2,"date":"2019-07-02 07:41:44.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-02 07:51:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-02 07:51:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 07:41:44.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-02 07:51:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-02 14:33:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48c53a7ba357c6511cdac53a6ced30c394cd95d4","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-29 07:40:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a876322606586fce7c557e9596fae9922ac4024","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 09:37:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/668287/1/check/openstack-ansible-linters/19e62ed/ : SUCCESS in 5m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_metal-debian-stable/6c39ee3/ : SUCCESS in 1h 03m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_metal-centos-7/5e122b2/ : SUCCESS in 1h 16m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/54bcb31/ : SUCCESS in 1h 21m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5c73296/ : SUCCESS in 1h 02m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/38179b6/ : SUCCESS in 1h 03m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/19defe1/ : SUCCESS in 56m 06s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/87/668287/1/check/openstack-tox-docs/4de900d/html/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d2ea9c0694a91606974367dd7f608fa41eb9f63c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-02 07:41:44.000000000","message":"Patch Set 1: Code-Review+2\n\nPasses CI.","accounts_in_message":[],"_revision_number":1},{"id":"1e36585297fe0150710d64eeaf244339227724a1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-02 07:51:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"702168b62f17898fbc2c1abb3c76955885c736e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 07:51:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee6076f8b9c204555b3561222c12f06b6597cb28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 09:41:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/87/668287/1/gate/openstack-ansible-linters/6288e10/ : SUCCESS in 6m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/d3a3b3d/ : FAILURE in 38m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_metal-centos-7/6c9e612/ : SUCCESS in 1h 42m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/0b204e3/ : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fc677c0/ : SUCCESS in 1h 02m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/3bccf18/ : SUCCESS in 1h 15m 59s\n- openstack-tox-docs http://logs.openstack.org/87/668287/1/gate/openstack-tox-docs/d3379f6/html/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5baa424e3a2bd2e2de7ccc32d6238b2242f403cd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-02 11:24:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3ee5d091a068f71c14d22824a0fa9edeac052f67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 13:01:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/668287/1/check/openstack-ansible-linters/9ca55a5/ : SUCCESS in 8m 18s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_metal-debian-stable/ac20791/ : SUCCESS in 1h 06m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_metal-centos-7/d6238b6/ : SUCCESS in 1h 28m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/7ad2a5a/ : SUCCESS in 1h 28m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/70a8729/ : SUCCESS in 57m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/a7c165a/ : SUCCESS in 1h 08m 57s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/87/668287/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/baf43f5/ : SUCCESS in 50m 49s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/87/668287/1/check/openstack-tox-docs/e7a1db5/html/ : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ca31c6024eabdc4517ae6d3b088f1231a0f49338","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 13:01:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d7e60e4a6092c7de15a2c6da9e91f7c24dcb401","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 14:33:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/668287/1/gate/openstack-ansible-linters/691d95e/ : SUCCESS in 4m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/24c358a/ : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_metal-centos-7/24d122c/ : SUCCESS in 1h 26m 27s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/a135074/ : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f841075/ : SUCCESS in 1h 13m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/87/668287/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/783d651/ : SUCCESS in 1h 04m 15s\n- openstack-tox-docs http://logs.openstack.org/87/668287/1/gate/openstack-tox-docs/4205119/html/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5ac310cc6358e5e95e01810624ebcf3fe6fc7a0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 14:33:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc04fc27bfaf45670a85143be7517b2de5cf7210","revisions":{"dc04fc27bfaf45670a85143be7517b2de5cf7210":{"kind":"REWORK","_number":1,"created":"2019-06-29 07:40:59.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/87/668287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/87/668287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/87/668287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/87/668287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/87/668287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/87/668287/1"}}},"commit":{"parents":[{"commit":"d26fda9287dfb23964313b03241650512c1d1180","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d26fda9287dfb23964313b03241650512c1d1180"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-29 07:39:36.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-29 07:40:51.000000000","tz":120},"subject":"Fix distro package lists for openSUSE 15","message":"Fix distro package lists for openSUSE 15\n\nExplicitly pulling in packages that are dependencies of the main\npackage can cause issues with newer releases where those packages\ndon\u0027t exist anymore - e.g. python-swift is renamed to python3-swift.\nsimilarly python-keystonemiddleware (which is a dep of openstack-swift)\nis actually called python3-keystonemiddleware these days.\n\nChange-Id: Iff4011c5a6531fa3f74acfeb6fa980c493246a63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/dc04fc27bfaf45670a85143be7517b2de5cf7210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/dc04fc27bfaf45670a85143be7517b2de5cf7210"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
