)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~671329","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I7553c3ef4d7620844f614da79b2cb1086eda23cb","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"static_imports","hashtags":[],"change_id":"I7553c3ef4d7620844f614da79b2cb1086eda23cb","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-07-17 16:26:07.000000000","updated":"2019-07-19 12:39:28.000000000","submitted":"2019-07-19 01:03:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671329-1563498196795-02c3ebfc","meta_rev_id":"c12a096a720be47e122b68dc1aa1ceb05ee3ecd6","_number":671329,"virtual_id_number":671329,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-18 16:58:57.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 01:03:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-18 12:39:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-18 12:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-18 12:51:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 12:39:23.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-18 12:51:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-19 01:03:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a28c645d62c79101e2008deb2cd6a6fa557fbad0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-17 16:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a7a53aba831cea373d8bd2273fc5dc96e6957e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:25:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/29/671329/1/check/openstack-ansible-linters/94617ba/ : SUCCESS in 4m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-debian-stable/458d73e/ : SUCCESS in 1h 09m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-centos-7/de37090/ : SUCCESS in 1h 45m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/dbbcb1d/ : SUCCESS in 1h 33m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a3c23eb/ : SUCCESS in 1h 27m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/7c2d66c/ : SUCCESS in 1h 09m 58s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/8a4044a/ : FAILURE in 49m 06s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/29/671329/1/check/openstack-tox-docs/02a3d14/html/ : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c616e5891ba4c77790f91c3430b49fd386643a44","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-18 12:39:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ee9a4d676037e72adc38c6b95c0ff999f15d90d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-18 12:51:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f40a7bcba68d1d341a6dffa98161a77a3dadd92a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 12:52:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"714bd08be54a32cf79baf223648450b93fc56e83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:33:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/29/671329/1/gate/openstack-ansible-linters/f445a24/ : SUCCESS in 6m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/f517ce9/ : SUCCESS in 1h 06m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-centos-7/1606b8b/ : SUCCESS in 1h 26m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/f783254/ : SUCCESS in 1h 35m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/513955b/ : FAILURE in 37m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/5e7a8bc/ : SUCCESS in 1h 05m 53s\n- openstack-tox-docs http://logs.openstack.org/29/671329/1/gate/openstack-tox-docs/6b50cc4/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"528e43edd52e55258e81b43ae54c69a1c7162304","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-18 16:58:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"278619d4cfc2b8940e7948692d81975cadcac9bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 18:41:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/29/671329/1/check/openstack-ansible-linters/928a100/ : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-debian-stable/be8343e/ : SUCCESS in 1h 19m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-centos-7/0c831dd/ : SUCCESS in 1h 33m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/255ea2f/ : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6ee364f/ : SUCCESS in 1h 19m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e71b925/ : SUCCESS in 1h 15m 58s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/82c11ff/ : FAILURE in 1h 01m 48s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/29/671329/1/check/openstack-tox-docs/ed2b096/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d3b8a1005123c40616b78e9ea9e1b1ba53049e9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 18:41:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a48f7728ecdb394a13db180ba3a71d5d33c332a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 21:32:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/29/671329/1/gate/openstack-ansible-linters/a1903a8/ : SUCCESS in 6m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/9306837/ : SUCCESS in 1h 32m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-centos-7/a425b4d/ : SUCCESS in 1h 50m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/b7ce9a4/ : SUCCESS in 1h 37m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e5337d3/ : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/c931f40/ : FAILURE in 1h 30m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/b6df1a8/ : SUCCESS in 53m 48s\n- openstack-tox-docs http://logs.openstack.org/29/671329/1/gate/openstack-tox-docs/29cbaf5/html/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d82eee13013fbbd9e4924c2c89b996e16375aa83","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-18 21:40:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c349d937c20f18654ac9759ae1f773affd6cb99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 23:29:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/29/671329/1/check/openstack-ansible-linters/e0bddb8/ : SUCCESS in 5m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-debian-stable/1368a4a/ : SUCCESS in 1h 27m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-centos-7/9a9d206/ : SUCCESS in 1h 43m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/beb8983/ : SUCCESS in 1h 44m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b717efc/ : SUCCESS in 1h 02m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5ae6dcf/ : SUCCESS in 1h 07m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/bf946d4/ : FAILURE in 1h 07m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/29/671329/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/8cf2005/ : SUCCESS in 53m 51s\n- openstack-tox-docs http://logs.openstack.org/29/671329/1/check/openstack-tox-docs/7e86216/html/ : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e08e674cdbd81c635aec2e30f5e0fdb045cc5e46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 23:29:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f91c9f66296f7a8728e38761400724ec566bdc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 01:03:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0dd557fa46a0824e5eaf2f8d9289a40735942dc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 01:03:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/29/671329/1/gate/openstack-ansible-linters/96115a2/ : SUCCESS in 7m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/bd967ca/ : SUCCESS in 1h 04m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-centos-7/b208680/ : SUCCESS in 1h 28m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/7e10fec/ : SUCCESS in 1h 03m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/1046713/ : SUCCESS in 1h 03m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/f5efc0f/ : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/29/671329/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/485b7b2/ : SUCCESS in 48m 16s\n- openstack-tox-docs http://logs.openstack.org/29/671329/1/gate/openstack-tox-docs/44066bd/html/ : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3be511c97d30838f2d081eef8a29d004aa81277b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 12:39:28.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 2cc65580535e12a2115f98dd2b303ae9bc6d7bcf","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"273895816b5e4846343be0e534af9501e2dfe000","revisions":{"273895816b5e4846343be0e534af9501e2dfe000":{"kind":"REWORK","_number":1,"created":"2019-07-17 16:26:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/671329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/29/671329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/29/671329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/29/671329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/29/671329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/29/671329/1"}}},"commit":{"parents":[{"commit":"6d64e4d0eb27eb03829a53dd5b6d7f17d96771b6","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/6d64e4d0eb27eb03829a53dd5b6d7f17d96771b6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 16:25:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 16:26:04.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like swift-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I7553c3ef4d7620844f614da79b2cb1086eda23cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/273895816b5e4846343be0e534af9501e2dfe000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/273895816b5e4846343be0e534af9501e2dfe000"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
