)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~671711","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I6dd0d77004394bb1ad674b53538b0679b056bb0f","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"use_journalctl","hashtags":[],"change_id":"I6dd0d77004394bb1ad674b53538b0679b056bb0f","subject":"Use systemd-journald instead of log files","status":"MERGED","created":"2019-07-19 12:11:19.000000000","updated":"2019-07-19 18:04:17.000000000","submitted":"2019-07-19 18:04:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671711-1563559457875-f7d455a8","meta_rev_id":"b961e060a619bf4a6e4dff50e4e6fb2aa553e737","_number":671711,"virtual_id_number":671711,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 18:04:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-19 16:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-19 16:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"date":"2019-07-19 12:11:33.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-19 16:30:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 16:10:26.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 16:30:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-19 18:04:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"123a7594dd82514282c3a1a7d27f70981f901ca3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 12:11:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12f8ae3dd9e780fb8dcfe3d386b11f2bdcfe20ce","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 12:11:33.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1879927e2134d4b73902790c65fde5f8df4810d2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 13:30:17.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"f60e93f040c7babf50936237847ca08a2c5480c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 13:48:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/11/671711/1/check/openstack-ansible-linters/bf9b953/ : SUCCESS in 6m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/11/671711/1/check/openstack-ansible-deploy-aio_metal-debian-stable/7c0bbd2/ : SUCCESS in 1h 11m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/11/671711/1/check/openstack-ansible-deploy-aio_metal-centos-7/85713d0/ : SUCCESS in 1h 32m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/11/671711/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/68623e7/ : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/11/671711/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/33e0e41/ : SUCCESS in 1h 04m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/11/671711/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5ffd477/ : SUCCESS in 1h 07m 42s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/11/671711/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/6664793/ : FAILURE in 51m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/11/671711/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/eb63f7f/ : SUCCESS in 48m 23s\n- openstack-tox-docs http://logs.openstack.org/11/671711/1/check/openstack-tox-docs/3a22457/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"89bf8b7598ee308c172c9466c57e4f811e509903","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 16:10:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a61bf8b1e05393b61ce68d8e0be1b97b54ca50b9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-19 16:30:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a14402c777c585859f311bc3b902c29eb6dca09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 16:30:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c936b6b288398bcf5b289a4080da2da6ffd240ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:04:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2cf13daecac250f1fb3997742017115fee281c89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:04:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/11/671711/1/gate/openstack-ansible-linters/fdb840f/ : SUCCESS in 7m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/11/671711/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/d49103a/ : SUCCESS in 1h 04m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/11/671711/1/gate/openstack-ansible-deploy-aio_metal-centos-7/e848eda/ : SUCCESS in 1h 25m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/11/671711/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/3970b6e/ : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/11/671711/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/72e860e/ : SUCCESS in 1h 09m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/11/671711/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/09021ed/ : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/11/671711/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/59c0916/ : SUCCESS in 56m 54s\n- openstack-tox-docs http://logs.openstack.org/11/671711/1/gate/openstack-tox-docs/4e28c58/html/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0fac1b559ade15cc6d66e39168da45b99be4a9a","revisions":{"d0fac1b559ade15cc6d66e39168da45b99be4a9a":{"kind":"REWORK","_number":1,"created":"2019-07-19 12:11:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/671711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/11/671711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/11/671711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/11/671711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/11/671711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/11/671711/1"}}},"commit":{"parents":[{"commit":"273895816b5e4846343be0e534af9501e2dfe000","subject":"Convert dynamic includes to static imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/273895816b5e4846343be0e534af9501e2dfe000"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-19 12:10:46.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-19 12:10:49.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald.\nBy this we mean dropping rsyslog client installation. log_address is set\nby default to /dev/log, which is served by journald.\n\nChange-Id: I6dd0d77004394bb1ad674b53538b0679b056bb0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d0fac1b559ade15cc6d66e39168da45b99be4a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d0fac1b559ade15cc6d66e39168da45b99be4a9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
