)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~671716","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Fstein~I7553c3ef4d7620844f614da79b2cb1086eda23cb","project":"openstack/openstack-ansible-os_swift","branch":"stable/stein","topic":"static_imports-stable/stein","hashtags":[],"change_id":"I7553c3ef4d7620844f614da79b2cb1086eda23cb","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-07-19 12:39:28.000000000","updated":"2019-07-19 18:48:03.000000000","submitted":"2019-07-19 18:48:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671716-1563562083114-77280bbb","meta_rev_id":"84bc7ffb11eddd67ddb6efb068ab96d2f3bb145e","_number":671716,"virtual_id_number":671716,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 18:48:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-19 16:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-19 16:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-19 16:30:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 16:09:35.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 16:30:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-19 18:48:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b3737b23e61c2dbc032ec30a28f292f67cf6f6c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 12:39:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1e11aa916e154461c1849913ce93a2c1d4b38fd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 14:02:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/16/671716/1/check/openstack-ansible-linters/7165c07/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/16/671716/1/check/openstack-tox-docs/83345f8/html/ : SUCCESS in 2m 33s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/16/671716/1/check/openstack-ansible-deploy-aio_metal_swift-centos-7/48610d0/ : SUCCESS in 1h 21m 51s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/16/671716/1/check/openstack-ansible-deploy-aio_metal_swift-debian-stable/cf8f5ed/ : SUCCESS in 1h 00m 44s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/16/671716/1/check/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/4573728/ : SUCCESS in 56m 21s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/16/671716/1/check/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/f6f3f06/ : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"915e0b295c1a0cce034a17ad8ed427e569cc1211","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 16:09:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcffb0943f489e58c05835d460ab5a421db67793","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-19 16:30:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d44d34e904268b2d6dccc9cc62cd7ad06b9423a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 16:30:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19ebbac1d708e186af672f85a080c5b5841ee14f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:48:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/16/671716/1/gate/openstack-ansible-linters/e79feb2/ : SUCCESS in 7m 37s\n- openstack-tox-docs http://logs.openstack.org/16/671716/1/gate/openstack-tox-docs/a385785/html/ : SUCCESS in 3m 43s\n- openstack-ansible-deploy-aio_metal_swift-centos-7 http://logs.openstack.org/16/671716/1/gate/openstack-ansible-deploy-aio_metal_swift-centos-7/b54e2bb/ : SUCCESS in 1h 24m 31s\n- openstack-ansible-deploy-aio_metal_swift-debian-stable http://logs.openstack.org/16/671716/1/gate/openstack-ansible-deploy-aio_metal_swift-debian-stable/06022e7/ : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic http://logs.openstack.org/16/671716/1/gate/openstack-ansible-deploy-aio_metal_swift-ubuntu-bionic/a0d8417/ : SUCCESS in 1h 03m 43s\n- openstack-ansible-deploy-aio_distro_metal_swift-centos-7 http://logs.openstack.org/16/671716/1/gate/openstack-ansible-deploy-aio_distro_metal_swift-centos-7/0efb4a9/ : SUCCESS in 1h 04m 34s","accounts_in_message":[],"_revision_number":1},{"id":"68ac7aa90ea361ce3b389339154105c75c774296","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:48:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cc65580535e12a2115f98dd2b303ae9bc6d7bcf","revisions":{"2cc65580535e12a2115f98dd2b303ae9bc6d7bcf":{"kind":"REWORK","_number":1,"created":"2019-07-19 12:39:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/671716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/16/671716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/16/671716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/16/671716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/16/671716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/16/671716/1"}}},"commit":{"parents":[{"commit":"0fc8081b9ca69d4a2581e5c3b4eadf129974cc81","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/0fc8081b9ca69d4a2581e5c3b4eadf129974cc81"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 16:25:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-07-19 12:39:28.000000000","tz":0},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like swift-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I7553c3ef4d7620844f614da79b2cb1086eda23cb\n(cherry picked from commit 273895816b5e4846343be0e534af9501e2dfe000)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2cc65580535e12a2115f98dd2b303ae9bc6d7bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2cc65580535e12a2115f98dd2b303ae9bc6d7bcf"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
