)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~673655","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Ie8feb2e0a0d229aed75461257199bc4af20abbfc","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ie8feb2e0a0d229aed75461257199bc4af20abbfc","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-30 19:44:03.000000000","updated":"2019-07-30 20:18:55.000000000","submitted":"2019-07-30 20:18:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673655-1564517935153-622e802a","meta_rev_id":"274427367c1cc69766a08118c350fbe146bfc154","_number":673655,"virtual_id_number":673655,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-30 20:18:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-07-30 20:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-07-30 20:08:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 20:08:32.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-30 20:18:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb0a5d1b9161e83b52dc9e2e5577c84d6fe268aa","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-30 19:44:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"296bee1a63febe917b9010e7b65754870b40a82d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:06:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/55/673655/1/check/requirements-check/cdc071e/ : SUCCESS in 2m 37s\n- openstack-ansible-linters http://logs.openstack.org/55/673655/1/check/openstack-ansible-linters/cc4a439/ : SUCCESS in 6m 34s\n- openstack-tox-docs http://logs.openstack.org/55/673655/1/check/openstack-tox-docs/f61b7f1/html/ : SUCCESS in 4m 01s\n- build-openstack-releasenotes http://logs.openstack.org/55/673655/1/check/build-openstack-releasenotes/d497015/html/ : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"eccc12d014f4535b1e4a192f571465856275d238","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 20:08:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7dda82bdbbf97991c508b952952ab25a38fff699","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:08:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f071ced2d1df0eaf493a4883550aecc0f8b03be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:18:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/55/673655/1/gate/requirements-check/1a0f814/ : SUCCESS in 3m 23s\n- openstack-ansible-linters http://logs.openstack.org/55/673655/1/gate/openstack-ansible-linters/db84f2c/ : SUCCESS in 7m 05s\n- openstack-tox-docs http://logs.openstack.org/55/673655/1/gate/openstack-tox-docs/c92cad6/html/ : SUCCESS in 4m 05s\n- build-openstack-releasenotes http://logs.openstack.org/55/673655/1/gate/build-openstack-releasenotes/15058fc/html/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"252e4d38081fef3a798c447eed2fbdd32f4dd98b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:18:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94870a6c67e0cd3fdd9b3695d47e0b261523fe38","revisions":{"94870a6c67e0cd3fdd9b3695d47e0b261523fe38":{"kind":"REWORK","_number":1,"created":"2019-07-30 19:44:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/673655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/55/673655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/55/673655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/55/673655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/55/673655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/55/673655/1"}}},"commit":{"parents":[{"commit":"e90a9a4d8ba706629e515156270c9babe7d81dc0","subject":"Merge \"service_setup: refactor service setup to a single file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/e90a9a4d8ba706629e515156270c9babe7d81dc0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:43:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:43:59.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie8feb2e0a0d229aed75461257199bc4af20abbfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/94870a6c67e0cd3fdd9b3695d47e0b261523fe38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/94870a6c67e0cd3fdd9b3695d47e0b261523fe38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
