)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~680946","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Ic4b476dd6794bd536bd31e6c1927fa65d1f87523","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"osa-py3","hashtags":[],"change_id":"Ic4b476dd6794bd536bd31e6c1927fa65d1f87523","subject":"[DNM] test swift / py3","status":"ABANDONED","created":"2019-09-09 08:37:54.000000000","updated":"2020-06-03 07:58:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ec4b3566618b929251067fd3a92a808ff3e9be5a","_number":680946,"virtual_id_number":680946,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-25 13:37:59.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":-1,"date":"2019-09-17 16:14:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 16:14:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-25 13:37:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"e036c144041258249a2929269766d94164ba3391","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-09 08:37:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2bd469d7a7bcba3f92ac54d26737f75b84fd7fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 10:04:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eddd1df8c870410e9a53ad8296d7b3ec : SUCCESS in 7m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/816bf9a224744060b440a31a9863a2b4 : FAILURE in 1h 06m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/25c471558d3144c08f5803b46f0d83e4 : FAILURE in 31m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ae7c6e75cfb749de89039d7c65291c8c : FAILURE in 1h 01m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/420943e5ae8246eb94ab517f7f7975c4 : FAILURE in 1h 02m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/47eb6f5fdebf4d0ca2a642bc38bc1359 : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c5b2621bc8f54f0898fb0e45008a41bb : FAILURE in 28m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/123281db2d9f4311865ad58f840af4a2 : SUCCESS in 51m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed2f4b5dcd24e2fbb0fc3f4466b2d01 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1faaef0144cf8be243f896da8b29e41f75fe410d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-09 11:00:26.000000000","message":"Topic set to osa-py3","accounts_in_message":[],"_revision_number":1},{"id":"5663b649522de96888c77dbbd48fd12153855f6b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-10 16:30:09.000000000","message":"Patch Set 1:\n\nFWIW, that\n\n Error: failed to load build file \u0027account.builder\u0027: \u0027utf-8\u0027 codec can\u0027t decode byte 0x80 in position 0: invalid start byte\n\nerror is because swift_rings_check.py opens the builder file in text mode on py3; something like\n\ndiff --git a/templates/swift_rings_check.py.j2 b/templates/swift_rings_check.py.j2\nindex 2c7a721..a2f7185 100644\n--- a/templates/swift_rings_check.py.j2\n+++ b/templates/swift_rings_check.py.j2\n@@ -36,7 +36,7 @@ def get_build_file_data(build_file):\n     build_file_data \u003d None\n     if exists(build_file):\n         try:\n-            with open(build_file) as bf_stream:\n+            with open(build_file, \u0027rb\u0027) as bf_stream:\n                 build_file_data \u003d pickle.load(bf_stream)\n         except Exception as ex:\n             print(\"Error: failed to load build file \u0027%s\u0027: %s\" % (build_file,\n\nshould help.","accounts_in_message":[],"_revision_number":1},{"id":"1d7a016a5587e5cd24cd166712ac298ba9aaf9ed","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-11 07:44:21.000000000","message":"Patch Set 1:\n\n\u003e FWIW, that\n \u003e \n \u003e Error: failed to load build file \u0027account.builder\u0027: \u0027utf-8\u0027 codec\n \u003e can\u0027t decode byte 0x80 in position 0: invalid start byte\n \u003e \n \u003e error is because swift_rings_check.py opens the builder file in\n \u003e text mode on py3; something like\n \u003e \n \u003e diff --git a/templates/swift_rings_check.py.j2 b/templates/swift_rings_check.py.j2\n \u003e index 2c7a721..a2f7185 100644\n \u003e --- a/templates/swift_rings_check.py.j2\n \u003e +++ b/templates/swift_rings_check.py.j2\n \u003e @@ -36,7 +36,7 @@ def get_build_file_data(build_file):\n \u003e build_file_data \u003d None\n \u003e if exists(build_file):\n \u003e try:\n \u003e -            with open(build_file) as bf_stream:\n \u003e +            with open(build_file, \u0027rb\u0027) as bf_stream:\n \u003e build_file_data \u003d pickle.load(bf_stream)\n \u003e except Exception as ex:\n \u003e print(\"Error: failed to load build file \u0027%s\u0027: %s\" % (build_file,\n \u003e \n \u003e should help.\n\nExcellent thanks for the tip! https://review.opendev.org/681396","accounts_in_message":[],"_revision_number":1},{"id":"c20c025ecfc7868c64500312de77dd7b44927737","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-11 07:46:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b192dab618714dd4a38944d30f7c826062a8f353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:05:37.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"fa0cbbdc363e02d75746ad909ae3fb79c7733eba","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-14 19:35:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"636bf9f0f53c0985c905070862ac7d38ad6de787","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-17 10:48:39.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"74f4d39330babb93f4dbbd524771c3b2d294f683","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-17 10:48:53.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"495f194b6a45fe87cfefb40ad57aab3586a153d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:39:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/27c398765abf4a6ea3cca4b3652b111f : SUCCESS in 6m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9913e96744f347afbe3a17c1685fd7b7 : FAILURE in 1h 21m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d3f73b2994854ba8b8f529f14c81f001 : SUCCESS in 1h 44m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aa4af1c6c768406891c3d5de55071914 : FAILURE in 1h 33m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60dad9dd6d8541e6a42ec56f4dc17ec8 : FAILURE in 1h 22m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7f142e3581ea4d408abc505d20fe5774 : FAILURE in 31m 02s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6d1757172f6f4785beb461628b1ee6a6 : FAILURE in 38m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1c96d0ba7e44a8186bfb6e55d96a4d7 : SUCCESS in 55m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd1916e90f41403c8e866b829fcff158 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"8fd4887afa26c06d7bc15078dde2176c14cf011a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-17 14:07:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"693c1eabfad201a60897191662f82b01b9bf5357","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:14:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/648f5cb1237747098c616b8104ce3145 : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/34668e04bc2245f9847666525e13f090 : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/01a3ab398f2640778f036fa5dba55220 : FAILURE in 2h 03m 28s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e2e3f1b0762540e0823193838f12d7f0 : SUCCESS in 1h 37m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20c7cf519ccc410aa7253aa81f83dce1 : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/350864e07f5f461c98a0a97e916508d7 : FAILURE in 30m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5cee619a46924c3aad010e4570e74a32 : FAILURE in 30m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98eddf8eecbd4a6c9fa770ebadf29168 : FAILURE in 1h 23m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2994aaf68324130b9fcd5c0bd8f82b5 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":4},{"id":"814ec9cf6e8976ec620459703c6d0e1651c250e8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-25 13:37:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"efeb813dc27484405115dd3b1b3cb1cf7393d19e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 16:49:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4f50a13cd54944a384de0afe9d8e73b9 : SUCCESS in 5m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/22fe79cec45b4d42a4a1794ebe3a9d2e : FAILURE in 1h 07m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cced40e490a94402982dce077fe38ede : FAILURE in 1h 42m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fa67b204790e48a8a6112c88ad348d8a : FAILURE in 1h 29m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7997c4fccd384aa986c7aa1222bd8c0f : FAILURE in 1h 13m 52s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3c2cf60320af46068013e160b989638b : SUCCESS in 2h 39m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e71a2fe2fb60455d99768c9124102f2d : SUCCESS in 1h 48m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df5c268511904241abac8e7bd0eeec1a : FAILURE in 1h 16m 18s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a6a908dee13f41cbb4750116f1ff0f48 : FAILURE in 1h 11m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3fa32bd084946439bce0996b8d6b83d : FAILURE in 59m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bf1d9eb034b4f57aef2b848fcf382a7 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":4},{"id":"ec4b3566618b929251067fd3a92a808ff3e9be5a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-03 07:58:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c1a011f86e6b48231051cfc4895988be8524aef9","revisions":{"b8ee6db57446f8f75a35974cb5d613bfcc0d042f":{"kind":"REWORK","_number":1,"created":"2019-09-09 08:37:54.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/46/680946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/46/680946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/1"}}},"commit":{"parents":[{"commit":"7754168dbb4d903400a2f07cd888258e032eabce","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/7754168dbb4d903400a2f07cd888258e032eabce"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-09 08:34:14.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-09 08:34:14.000000000","tz":60},"subject":"[DNM] test swift / py3","message":"[DNM] test swift / py3\n\ndepends-on: https://review.opendev.org/680944\ndepends-on: https://review.opendev.org/680470\nChange-Id: Ic4b476dd6794bd536bd31e6c1927fa65d1f87523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/b8ee6db57446f8f75a35974cb5d613bfcc0d042f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/b8ee6db57446f8f75a35974cb5d613bfcc0d042f"}]},"branch":"refs/heads/master"},"676cf95876e2fd36bf3ec3b5bb383c6ed4070223":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-11 07:46:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/46/680946/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/46/680946/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/2"}}},"commit":{"parents":[{"commit":"7754168dbb4d903400a2f07cd888258e032eabce","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/7754168dbb4d903400a2f07cd888258e032eabce"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-09 08:34:14.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-11 07:44:48.000000000","tz":0},"subject":"[DNM] test swift / py3","message":"[DNM] test swift / py3\n\ndepends-on: https://review.opendev.org/680944\ndepends-on: https://review.opendev.org/680470\ndepends-on: https://review.opendev.org/681396\nChange-Id: Ic4b476dd6794bd536bd31e6c1927fa65d1f87523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/676cf95876e2fd36bf3ec3b5bb383c6ed4070223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/676cf95876e2fd36bf3ec3b5bb383c6ed4070223"}]},"branch":"refs/heads/master"},"c37e702577cec2d7c96acc2e1d3239d4223970b7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-17 10:48:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/680946/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/46/680946/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/3"}}},"commit":{"parents":[{"commit":"7754168dbb4d903400a2f07cd888258e032eabce","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/7754168dbb4d903400a2f07cd888258e032eabce"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-09 08:34:14.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-09-17 10:48:50.000000000","tz":0},"subject":"[DNM] test swift / py3","message":"[DNM] test swift / py3\n\n\nChange-Id: Ic4b476dd6794bd536bd31e6c1927fa65d1f87523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c37e702577cec2d7c96acc2e1d3239d4223970b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c37e702577cec2d7c96acc2e1d3239d4223970b7"}]},"branch":"refs/heads/master"},"c1a011f86e6b48231051cfc4895988be8524aef9":{"kind":"REWORK","_number":4,"created":"2019-09-17 14:07:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/680946/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/46/680946/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/46/680946/4"}}},"commit":{"parents":[{"commit":"25d0812be561d17c1400110d6dd6025150438639","subject":"Merge \"Update swift_rings_check script for python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/25d0812be561d17c1400110d6dd6025150438639"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-17 14:07:12.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-17 14:07:12.000000000","tz":180},"subject":"[DNM] test swift / py3","message":"[DNM] test swift / py3\n\nChange-Id: Ic4b476dd6794bd536bd31e6c1927fa65d1f87523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c1a011f86e6b48231051cfc4895988be8524aef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c1a011f86e6b48231051cfc4895988be8524aef9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
