)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~682827","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Ib4032128a5b44a1be3d7e4927f4e807406b36dc9","project":"openstack/openstack-ansible-os_swift","branch":"master","hashtags":[],"change_id":"Ib4032128a5b44a1be3d7e4927f4e807406b36dc9","subject":"Temporary fix for python3 config parsing error","status":"ABANDONED","created":"2019-09-18 08:14:43.000000000","updated":"2019-09-18 19:18:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee32d00251ec918c0f878056860d9e5ffe19f583","_number":682827,"virtual_id_number":682827,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-09-18 10:09:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 10:09:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41635b2d6c6df6a17fbd5b24c826046abafce025","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-18 08:14:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17a77361dd70f33f22b6611f48e55b11f2d13b0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:09:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/54384147eb8743d2bd0561244ccf891e : SUCCESS in 7m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3285a0106c654067b0b5b71840953025 : SUCCESS in 1h 31m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1b5bb6aa0d7c4aef86b246676b6a49c1 : FAILURE in 1h 47m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3e96b260555e4923bc126130ca5c13a6 : SUCCESS in 1h 35m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d20190c3067249bba72bcc9adf2b51c0 : SUCCESS in 1h 12m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1df10d00dfae4412886359c8f589fef9 : FAILURE in 39m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e28dc72bb2c84bdeba5a916a4a345914 : FAILURE in 32m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df20425bf1e949f3a1321d6d9f2c5564 : FAILURE in 58m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a449e5f591e402b9e700fb9e0242e69 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ee32d00251ec918c0f878056860d9e5ffe19f583","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-18 19:18:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1aa3725536645f76af558789589b32c94504fe18","revisions":{"1aa3725536645f76af558789589b32c94504fe18":{"kind":"REWORK","_number":1,"created":"2019-09-18 08:14:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/27/682827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/27/682827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/27/682827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/27/682827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/27/682827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/27/682827/1"}}},"commit":{"parents":[{"commit":"25d0812be561d17c1400110d6dd6025150438639","subject":"Merge \"Update swift_rings_check script for python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/25d0812be561d17c1400110d6dd6025150438639"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-18 08:11:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-18 08:13:10.000000000","tz":60},"subject":"Temporary fix for python3 config parsing error","message":"Temporary fix for python3 config parsing error\n\nRevert this patch when the bug in swift is fixed\n\nSee https://bugs.launchpad.net/swift/+bug/1844368\n\nChange-Id: Ib4032128a5b44a1be3d7e4927f4e807406b36dc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1aa3725536645f76af558789589b32c94504fe18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1aa3725536645f76af558789589b32c94504fe18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
