)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~682874","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I531d9fdf424d1d2dfaeb15e01d3e75d3e532505c","project":"openstack/openstack-ansible-os_swift","branch":"master","hashtags":[],"change_id":"I531d9fdf424d1d2dfaeb15e01d3e75d3e532505c","subject":"Update invaild link for Conf","status":"ABANDONED","created":"2019-09-18 11:51:57.000000000","updated":"2020-01-16 02:22:11.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92110ba71b0ac559fd9c4f70eed9a2d51d9d7ff6","_number":682874,"virtual_id_number":682874,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-05 02:27:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":1,"date":"2019-11-06 06:27:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-11-08 20:15:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-31 09:28:34.000000000","updated_by":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"reviewer":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"state":"REVIEWER"},{"updated":"2019-11-05 02:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-06 06:27:10.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2019-11-08 20:15:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"72f95f1b01b1e728f3d90c4fe4e6dddf487fbd08","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-18 11:51:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa3b40f74a5710a6cef125f965452b408a715a85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:03:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d372dbb2e48740c397c9d73b9a5b842e : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74be867e8beb4ad09a4e8344f44124c3 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/286c6077d1d049cfb36b8339590165f9 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"12f4393578738769915d3eefc5c7fe1dcc290604","author":{"_account_id":27383,"name":"Shuo Liu","email":"liushuobj@inspur.com","username":"liushuobj"},"date":"2019-10-31 09:28:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"827e2144d9c81854188e28f9b711d082f0b12be1","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-11-05 02:18:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6e8f2c01f0fc1e737bf14ee542bcdaca6fd1d0d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 02:18:54.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"868f112a991aae9a6ab75615a1b3684a4b85acbd","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-11-05 02:19:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f121ca9349c7315d9da48fc3c50ed70dea6f5dc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 02:27:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/34d8ffe9ef7845708cc0b17e1f7633a0 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afbc338129354b65a6e257dc0f99c0c2 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54f86425d0144a91b082f098b9b454a3 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":3},{"id":"9875461133bb73165be05c937de1485e477a8de1","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2019-11-06 06:27:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"13dd887d67c1eddcb6d4089fbd80f14870a25816","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-08 20:15:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"92110ba71b0ac559fd9c4f70eed9a2d51d9d7ff6","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-01-16 02:22:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d93a5010bebcd92129b8359dd2ecc931ccdd0b9c","revisions":{"65f20be8e3aad19ae9d375e64a5f696c26809236":{"kind":"REWORK","_number":1,"created":"2019-09-18 11:51:57.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/74/682874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/74/682874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/1"}}},"commit":{"parents":[{"commit":"25d0812be561d17c1400110d6dd6025150438639","subject":"Merge \"Update swift_rings_check script for python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/25d0812be561d17c1400110d6dd6025150438639"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-18 11:51:45.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-18 11:51:45.000000000","tz":480},"subject":"Update invaild link for Conf","message":"Update invaild link for Conf\n\nChange-Id: I531d9fdf424d1d2dfaeb15e01d3e75d3e532505c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/65f20be8e3aad19ae9d375e64a5f696c26809236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/65f20be8e3aad19ae9d375e64a5f696c26809236"}]},"branch":"refs/heads/master"},"16df89d68b6af29572dfca7a539d07231d2fcccf":{"kind":"REWORK","_number":2,"created":"2019-11-05 02:18:43.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/74/682874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/74/682874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/2"}}},"commit":{"parents":[{"commit":"25d0812be561d17c1400110d6dd6025150438639","subject":"Merge \"Update swift_rings_check script for python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/25d0812be561d17c1400110d6dd6025150438639"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-18 11:51:45.000000000","tz":480},"committer":{"name":"Qitao","email":"qihongtao@inspur.com","date":"2019-11-05 02:18:28.000000000","tz":0},"subject":"Update invaild link for Conf","message":"Update invaild link for Conf\n\nChange-Id: I531d9fdf424d1d2dfaeb15e01d3e75d3e532505c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/16df89d68b6af29572dfca7a539d07231d2fcccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/16df89d68b6af29572dfca7a539d07231d2fcccf"}]},"branch":"refs/heads/master"},"d93a5010bebcd92129b8359dd2ecc931ccdd0b9c":{"kind":"REWORK","_number":3,"created":"2019-11-05 02:19:41.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/74/682874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/74/682874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/682874/3"}}},"commit":{"parents":[{"commit":"25d0812be561d17c1400110d6dd6025150438639","subject":"Merge \"Update swift_rings_check script for python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/25d0812be561d17c1400110d6dd6025150438639"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-18 11:51:45.000000000","tz":480},"committer":{"name":"Qitao","email":"qihongtao@inspur.com","date":"2019-11-05 02:19:31.000000000","tz":0},"subject":"Update invaild link for Conf","message":"Update invaild link for Conf\n\nChange-Id: I531d9fdf424d1d2dfaeb15e01d3e75d3e532505c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d93a5010bebcd92129b8359dd2ecc931ccdd0b9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d93a5010bebcd92129b8359dd2ecc931ccdd0b9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
