)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~683788","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Fpike~I9d48d13086dfce0b8b98e99f42de0531d06f381f","project":"openstack/openstack-ansible-os_swift","branch":"stable/pike","topic":"goodbye_42.3","hashtags":[],"change_id":"I9d48d13086dfce0b8b98e99f42de0531d06f381f","subject":"Remove openSUSE 42.3 job","status":"MERGED","created":"2019-09-21 14:07:03.000000000","updated":"2019-10-14 16:31:12.000000000","submitted":"2019-10-14 16:31:12.000000000","submitter":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683788-1571070672481-33a7a8e1","meta_rev_id":"1f9f7ad66d47bf98309f6b20c815d7ececba8c4a","_number":683788,"virtual_id_number":683788,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-10-14 16:31:08.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":-1,"date":"2019-09-21 14:47:38.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2019-10-14 16:31:08.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2019-10-14 16:31:08.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"date":"2019-09-21 14:07:03.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-21 14:07:03.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-21 14:47:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-14 16:31:08.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"}],"messages":[{"id":"54eaf3c6a37e9e6c092dbe8a2ac4828a07fba062","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-21 14:07:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4df0827ef117eac702b844ba12107df40129d822","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 14:47:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c840790bf3ee4dc68d72a988dfc8fb40 : FAILURE in 8m 06s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0001902fee364ec7af57710e27094a75 : FAILURE in 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"64735bb2f86de5880b0dc1001e9d2f7700acdbd5","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-14 16:31:08.000000000","message":"Patch Set 1: Verified+2 Code-Review+2 Workflow+1\n\nThe opensuse-423 nodeset is being removed.  This change is being force-merged to correct the resulting configuration error.","accounts_in_message":[],"_revision_number":1},{"id":"afeec3c2985177a8407fec53b07f2d51d65b216f","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-10-14 16:31:12.000000000","message":"Change has been successfully merged by James E. Blair","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30155d70aa659e6177b79244f69e5839d1f6e550","revisions":{"30155d70aa659e6177b79244f69e5839d1f6e550":{"kind":"REWORK","_number":1,"created":"2019-09-21 14:07:03.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/88/683788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/88/683788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/88/683788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/88/683788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/88/683788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/88/683788/1"}}},"commit":{"parents":[{"commit":"8c7201b204288ff9edaee6f6889ac414f86b0311","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/8c7201b204288ff9edaee6f6889ac414f86b0311"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 11:08:06.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-21 14:06:48.000000000","tz":120},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove the job, we test with openSUSE 15 now (as\non master). This allows to remove the nodes from CI.\n\nChange-Id: I9d48d13086dfce0b8b98e99f42de0531d06f381f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/30155d70aa659e6177b79244f69e5839d1f6e550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/30155d70aa659e6177b79244f69e5839d1f6e550"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
