)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~688683","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Ic8a752ef71c1205362a4e88a6098d152ca1b307e","project":"openstack/openstack-ansible-os_swift","branch":"master","hashtags":[],"change_id":"Ic8a752ef71c1205362a4e88a6098d152ca1b307e","subject":"update source link in readme","status":"MERGED","created":"2019-10-15 11:31:48.000000000","updated":"2019-10-15 16:03:45.000000000","submitted":"2019-10-15 16:01:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688683-1571155280479-f9e06b29","meta_rev_id":"a34e6c2d4461fa0743371b65b7ac33be4a624ad6","_number":688683,"virtual_id_number":688683,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 16:01:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-10-15 12:14:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-15 15:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-15 15:50:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 12:14:24.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-15 15:50:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-15 16:01:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d81415cc49f634011bc71f6372387616840fd5fd","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-10-15 11:31:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e84e5d1c7276c4424b28a1ef982bdf6ffc3986b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 11:41:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4d6a11610ba74425817683210c5f78a1 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883a751ef5f0402797184031df0c3d25 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2cb51d9c861fe462c8e1f3148395ea9e759c2ac6","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-15 12:14:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0a5060e754cc1f650b53c07b8d9838dc6f97eaa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-15 15:50:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db3452b60b7edda9f46a17814882d086b2dd1652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 15:50:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c20f84f6212381c7fcf9ec398b13ec8dde6abb68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:01:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aa775a18383c53411085633587a8177e0f6179ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:01:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bb2edcab3da3446e8510bac03f24b731 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0149fde7884eb9a56fef39bc4b2f72 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a34e6c2d4461fa0743371b65b7ac33be4a624ad6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:03:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681d3a3d60774c3fb75b38e7ebc0241f : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87eb3dc3802c7e7bd5faf55ed50b7eefc4ba2606","revisions":{"87eb3dc3802c7e7bd5faf55ed50b7eefc4ba2606":{"kind":"REWORK","_number":1,"created":"2019-10-15 11:31:48.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/83/688683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/83/688683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/83/688683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/83/688683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/83/688683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/83/688683/1"}}},"commit":{"parents":[{"commit":"400e723ebe63682f4aacb1c2338ebd25a5413dd1","subject":"Merge \"Revert \"Update fallocate_reserve to work with py3\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/400e723ebe63682f4aacb1c2338ebd25a5413dd1"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-15 11:30:56.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-15 11:31:34.000000000","tz":480},"subject":"update source link in readme","message":"update source link in readme\n\nChange-Id: Ic8a752ef71c1205362a4e88a6098d152ca1b307e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/87eb3dc3802c7e7bd5faf55ed50b7eefc4ba2606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/87eb3dc3802c7e7bd5faf55ed50b7eefc4ba2606"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
