)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~705125","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I037a1bb60e303707b6ad2dc4429bdf03e49d154e","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"debian-libsystemd","hashtags":[],"change_id":"I037a1bb60e303707b6ad2dc4429bdf03e49d154e","subject":"add libsystemd-dev to swift_devel_distro_packages","status":"MERGED","created":"2020-01-30 22:13:28.000000000","updated":"2020-03-11 19:54:19.000000000","submitted":"2020-03-11 19:52:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705125-1583956366292-97979107","meta_rev_id":"039874ade3ba558a67920bdcc1e9e407571e240b","_number":705125,"virtual_id_number":705125,"owner":{"_account_id":31562,"name":"Henry Bonath","email":"henry@thebonaths.com","username":"hbonath"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 19:52:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-03-11 14:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-11 18:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-11 18:08:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-11 14:14:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-03-11 18:08:52.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-11 19:52:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71bd24362486fa7a96675af8c7d0a07bf5459b03","author":{"_account_id":31562,"name":"Henry Bonath","email":"henry@thebonaths.com","username":"hbonath"},"date":"2020-01-30 22:13:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1eb6701f8e3105f61fd868f6482a14ec08584ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 00:53:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/764bff106c924edb8d74aba3d76c710c : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3b2221b11c8e4761b137f665856a3f25 : SUCCESS in 1h 12m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/60197e7125f84c7abb8cd5742cbdb318 : SUCCESS in 1h 39m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e3cdd79a67aa41818fcaac80d189938e : SUCCESS in 1h 25m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39e3a78df1e0460aaf8d9f1e0b48752b : SUCCESS in 1h 12m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d6d659b9d3d34696bf70a03a56418810 : SUCCESS in 2h 38m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb4eb111ba3846b3961eecbc71e81ff3 : SUCCESS in 1h 56m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1df8a92d6c924d0bb77d10f0c532f171 : FAILURE in 52m 31s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5774982b77b148c4b07e7f0f8396fa2f : FAILURE in 28m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cb85402183e405081c78e01098188b1 : FAILURE in 46m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b42d44ae2f4ba8b9e51f88c6553a85 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"99c07eb5d1c742951173c26de73525cc7a43ce16","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-01-31 07:14:15.000000000","message":"Patch Set 1:\n\nok, tests failed for known issue with ubuntu on which jrosser works and something new for centos. Will take a closer look today.","accounts_in_message":[],"_revision_number":1},{"id":"f65d7858962a3ddd312c9241ebd583598c13d4fa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-10 22:25:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"69e9a3bbbbb5b4658df74e1952d784c722a7371e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 00:00:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/43d53a7efbe94ccf906903c1848178ed : POST_FAILURE in 9m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9d2e482206394c53a64648b7a7b5493b : SUCCESS in 1h 09m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/697fcbb3853546c2a7c49dbb5cf5fa92 : SUCCESS in 1h 31m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5d4fa67b8616424096bc974aa3d42644 : SUCCESS in 1h 29m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/623bdaf903324611a96975d8c9c8c375 : SUCCESS in 1h 24m 46s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/91603accdbda4189a26bcd4861a6f29f : FAILURE in 1h 18m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/359da1f6fbf540db9935510164752062 : FAILURE in 1h 18m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8ba2325f79804bc8932d7b846e988e1d : SUCCESS in 1h 20m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/eee8c095d0854abda39ba1a845eba34a : FAILURE in 33m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60e8ed2128f5493ba92e62f4e8ce54b9 : SUCCESS in 1h 04m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5cbfe7c1ce4fa5b92f9ba2553bee8d : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9432827b33f30f0a476f8b52b4de8c92f8909e9f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-11 09:47:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f4f2086bc5b92b53cea85fe4fbefdbb9d84992d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 11:49:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dcf256288ec342b6b5555daa5299379c : SUCCESS in 11m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f43c274248284a09b51b7b9de9354d9b : SUCCESS in 1h 08m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/39541a9de7834f64a3f13a90286c4feb : SUCCESS in 1h 46m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0b6ae572676e4725ab206a3290998c3a : SUCCESS in 1h 30m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a21ee0766f5843b9aef1fc313414f962 : SUCCESS in 1h 12m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5281c97f6d9d420187a73db32a52f82e : FAILURE in 2h 00m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27a6ade7c7954e03827a98ef2dc3170c : FAILURE in 1h 13m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fcbb14b4de034b07994ab3a09a3d4e2a : SUCCESS in 1h 24m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b10b3dd26d644ceca998578e74a116c1 : FAILURE in 42m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b0d378b6cb847268e4973727e9571b6 : SUCCESS in 58m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b8d327e34ad4814a54c6836b4e82fa0 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"86bcd2630430ca8e6bcf50f4900d3bccf9041c09","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-11 12:40:25.000000000","message":"Patch Set 2:\n\nHenry, could you kindly tell us more about this change - what issue does it fix, and why change only for debian/ubuntu is required (since I don\u0027t see similar package for rhel and suse)","accounts_in_message":[],"_revision_number":2},{"id":"3125eb70eb25d3a4da1a5a6d7e8668fdf3f6feaf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-11 12:41:05.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"056f5e502e49a49e409adce048fe908b297b3452","author":{"_account_id":31562,"name":"Henry Bonath","email":"henry@thebonaths.com","username":"hbonath"},"date":"2020-03-11 14:13:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"122121322f99a03d576b4754725d636345b6510b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-11 14:14:48.000000000","message":"Patch Set 3: Code-Review+2\n\nI guess the same may happen on buster, so LGTM","accounts_in_message":[],"_revision_number":3},{"id":"68bd786850bc4965c8f891a8746cedadcafb20a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:41:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7d476f8876164feea01d4c72796f130f : SUCCESS in 13m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dabaac086a9148a48477556cb83f49f4 : SUCCESS in 1h 10m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ccdaa074e23b4cb1899b493662705fee : SUCCESS in 2h 06m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/40051adf657c40c5b3ddf1b8035be3d1 : SUCCESS in 1h 46m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a466f1fcc5454289801b8b94bcac6d1f : SUCCESS in 1h 35m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e85fc3ac9464bd09888d3cf1cfc108e : TIMED_OUT in 3h 04m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/588c44c502ee46148638f72941b34a85 : SUCCESS in 2h 26m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/edcf76a214c348389a21845c027223e2 : SUCCESS in 1h 27m 27s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/29802b849f264dba964243e311b2d2c3 : FAILURE in 35m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f097031936d4894a1716221c888d0ff : SUCCESS in 1h 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd39428452945408896bb645d30e690 : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":3},{"id":"d155b5c4a5da8af1d0735278a951b5d7045c8a3b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-11 18:08:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ac97b1c8cbb6abeb798488226d5d01663c76e3d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 18:11:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6158998491521dc229844e888b6976391c4fa49e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 19:52:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/98bf087030f64293b86911a8b7ebdf42 : SUCCESS in 11m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/70f06113bbec416eac6cafd329767468 : SUCCESS in 1h 10m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6eadb8aed545419089068ea9d2273444 : SUCCESS in 1h 32m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cdd91a33c44844469660b820ed91214a : SUCCESS in 1h 36m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d40ad474ee29437cb924f7a0217929c6 : SUCCESS in 1h 19m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8b4d6f590adb45abbc7b8ec0fcd6e84a : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59cf4153c62941d1a669664ebf6d065f : SUCCESS in 1h 01m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66395799961d4301a452efde98472fab : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":3},{"id":"f7e92b4e035785390b805a89182fad048b967934","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 19:52:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"36275cd27c3e5991777c5244bf9d0525849d729f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 19:54:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9bad580450642c9b2fb867d0d8c5648 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2399cb1c8fcc18823d8d9772b60f1f1ec5476687","revisions":{"f59c070dc0acae41cc6bd7f1c7b24b277126c52e":{"kind":"REWORK","_number":1,"created":"2020-01-30 22:13:28.000000000","uploader":{"_account_id":31562,"name":"Henry Bonath","email":"henry@thebonaths.com","username":"hbonath"},"ref":"refs/changes/25/705125/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/25/705125/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/1"}}},"commit":{"parents":[{"commit":"1e6cc45d97dc92767fec9abb11c7d275f2f657d3","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1e6cc45d97dc92767fec9abb11c7d275f2f657d3"}]}],"author":{"name":"Henry Bonath","email":"henry@thebonaths.com","date":"2020-01-30 22:05:17.000000000","tz":-300},"committer":{"name":"Henry Bonath","email":"henry@thebonaths.com","date":"2020-01-30 22:13:27.000000000","tz":-300},"subject":"add libsystemd-dev to swift_devel_distro_packages","message":"add libsystemd-dev to swift_devel_distro_packages\n\nChange-Id: I037a1bb60e303707b6ad2dc4429bdf03e49d154e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/f59c070dc0acae41cc6bd7f1c7b24b277126c52e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/f59c070dc0acae41cc6bd7f1c7b24b277126c52e"}]},"branch":"refs/heads/master"},"fc19639299bc94cf312994a2ae07e85146da6ee3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-10 22:25:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/25/705125/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/25/705125/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/2"}}},"commit":{"parents":[{"commit":"1b7a0084cf5b03dd505f7fce8422e3966b3a752a","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1b7a0084cf5b03dd505f7fce8422e3966b3a752a"}]}],"author":{"name":"Henry Bonath","email":"henry@thebonaths.com","date":"2020-01-30 22:05:17.000000000","tz":-300},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-10 22:25:55.000000000","tz":0},"subject":"add libsystemd-dev to swift_devel_distro_packages","message":"add libsystemd-dev to swift_devel_distro_packages\n\nChange-Id: I037a1bb60e303707b6ad2dc4429bdf03e49d154e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fc19639299bc94cf312994a2ae07e85146da6ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fc19639299bc94cf312994a2ae07e85146da6ee3"}]},"branch":"refs/heads/master"},"2399cb1c8fcc18823d8d9772b60f1f1ec5476687":{"kind":"REWORK","_number":3,"created":"2020-03-11 14:13:46.000000000","uploader":{"_account_id":31562,"name":"Henry Bonath","email":"henry@thebonaths.com","username":"hbonath"},"ref":"refs/changes/25/705125/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/25/705125/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/25/705125/3"}}},"commit":{"parents":[{"commit":"1e6cc45d97dc92767fec9abb11c7d275f2f657d3","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1e6cc45d97dc92767fec9abb11c7d275f2f657d3"}]}],"author":{"name":"Henry Bonath","email":"henry@thebonaths.com","date":"2020-01-30 22:05:17.000000000","tz":-300},"committer":{"name":"Henry Bonath","email":"henry@thebonaths.com","date":"2020-03-11 14:10:16.000000000","tz":-240},"subject":"add libsystemd-dev to swift_devel_distro_packages","message":"add libsystemd-dev to swift_devel_distro_packages\n\nFix for Swift venv build failing on Debian 9 with \"Cannot find libsystemd or libsystemd-journal\"\n\nChange-Id: I037a1bb60e303707b6ad2dc4429bdf03e49d154e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2399cb1c8fcc18823d8d9772b60f1f1ec5476687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2399cb1c8fcc18823d8d9772b60f1f1ec5476687"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
