)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~705247","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~If42feb0bb3ed0de211e5fba65c59d9ef50d199ce","project":"openstack/openstack-ansible-os_swift","branch":"master","hashtags":[],"change_id":"If42feb0bb3ed0de211e5fba65c59d9ef50d199ce","subject":"Use py2 shebang for centos distro installs","status":"MERGED","created":"2020-01-31 16:03:56.000000000","updated":"2020-02-28 23:50:47.000000000","submitted":"2020-02-28 23:49:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705247-1582933742953-b16e9fe7","meta_rev_id":"f876bab85c6685e5722a792a1cc80fe63aa65b22","_number":705247,"virtual_id_number":705247,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"date":"2020-02-28 20:03:10.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 23:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-02-27 16:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-15 17:38:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2020-02-27 16:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-15 17:38:23.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-02-27 16:20:01.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-02-28 23:49:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76b40c102142543042c09cec0950111c5523a418","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-01-31 16:03:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9146ffb90ae4a001614d4f83e00d696251646ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:46:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/64153c4e5d5c42f595ce7422946c8c09 : SUCCESS in 6m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f97592ece51b4a98a45a03abf6114dc7 : SUCCESS in 1h 05m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cc77da894b1e48c39d50c3f503ccb1b3 : SUCCESS in 2h 02m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6955c0e0fd954fadb20e164ea5e620d4 : SUCCESS in 1h 57m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00a4a5af60e2421383e6e39f91a232cd : SUCCESS in 1h 11m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ecf11f75ad154cd48ab403e5c699f813 : SUCCESS in 2h 40m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/006540a72cfe499ba7d7a634568d1cf8 : SUCCESS in 1h 52m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/189680cb0ce84adf9c7b5d3e88f5f6b9 : FAILURE in 53m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/88acbf6f91d04f669993976fb1e1a32f : FAILURE in 31m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2873e10d2aeb41e781ea4e4b5f047537 : SUCCESS in 1h 01m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd71e9a9de94103b187a5f2fd61cf23 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3cf22acd216be2701256b8fd79a77ba25f449f79","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-11 07:14:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d88fa43c81dfe690ba75c05a7e0b811889f1a217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 08:38:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/85a7587aee154d879b384b4be1e9c98b : SUCCESS in 6m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cf508bfabb444e219690efd0677d6239 : FAILURE in 59m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ca9243802efb4d83b6788d8f88a5c5a5 : FAILURE in 19m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/eff80d5a9b584cf58a2ad495652e3bf0 : FAILURE in 1h 20m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec5ae393260c413cbacb4afcbc179cb8 : FAILURE in 17m 47s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e58013d5b7834349bd8312bc02592e06 : FAILURE in 22m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9fa645566144686b541001759b12b2a : FAILURE in 18m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c3e136c4cbe442499b48ee248642afb9 : FAILURE in 27m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/40a710da22b24f53ad3165c3c9c93668 : FAILURE in 30m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21c1c60df66d4621a95594da2cd87002 : FAILURE in 49m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6fdd0700b47459a880b159a85ccb94c : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"701dcf8b9b0bc93e9ff0ebab5d009c440eb5b9e3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-11 11:30:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"053d3eec6f155af503546ea0fe8a0303b042c0f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 12:57:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6859fc172d7447888b7ea5b9fd8767f3 : SUCCESS in 8m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/05a93722279c4135a4b9eb1d9ee441d8 : SUCCESS in 1h 11m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/08c3cfba012a4cf4948da399ba15707d : FAILURE in 18m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d1ecf1d6e1b64f0faa545a9428422544 : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18f2fbeda9a24e129edcfa957b03c5c5 : FAILURE in 15m 49s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b208d8609e0a4562b0952a4897563626 : FAILURE in 15m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d5bc01a92b44f2792d9ef87e7d302d5 : FAILURE in 17m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/021aabea56dd454d87c03d6a933969a9 : FAILURE in 33m 33s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/38fa396c2cbf420698c973875aca12fd : FAILURE in 34m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3873628c78545b597010c6a8d91f46d : FAILURE in 42m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2b0ed466c74334b0efc7b987518639 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"9e60eaabc25ed0e8a2e608c04c6eae014f9c5972","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-15 13:38:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"42fd2d8849b3d89368c76c8aa7cad0814474339a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-15 13:41:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"640e65950406a7e8fd1172c688d9994d15f46221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 16:31:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/97d0505e0c3c46b4974bdfaa899a9c0b : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e8390b3f10104f8891d664019dc869ff : SUCCESS in 1h 09m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/775268832bb24b8cbc6927c78180a411 : SUCCESS in 1h 37m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/007feeb319f746848ef05cae29a6a2b0 : SUCCESS in 1h 22m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98a7b8f0d91d4a10b6884f214fca5697 : SUCCESS in 1h 21m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/acd8c16b2a3c4293b62308d0923baf0a : SUCCESS in 2h 48m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38140a941cd740eb8e28d09a7e63f549 : SUCCESS in 1h 55m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d22a3dbde55143f086b94c9727cd847b : SUCCESS in 1h 17m 30s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4fcc49d1c6084d08a5513bebb54f9964 : FAILURE in 28m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe1b4815a8c946678e3b5b80a943a702 : SUCCESS in 56m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a16ad3abd7473bbbaf708eea1c8f35 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":3},{"id":"20d1668fdea5e1228448b6aa63d2645b54d46b9c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-15 17:38:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"01e43bf01257b79392a332c70b756552d78c892b","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-02-27 16:20:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"833ab988832dfbb2f0b3c74b923b7cdb2bd707e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:20:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fa38ef8855eaff33bbb46427f955df7769f15bd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 18:15:42.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/466582693da54b72b0823c764f8a616b : SUCCESS in 8m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/816582677021422793df186c9fb13574 : RETRY_LIMIT in 36m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/408af8c67fe842958f8fe84efe57a076 : RETRY_LIMIT in 36m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1d579300ed0e45578ddd919b4a25691b : RETRY_LIMIT in 35m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50f25e70b79d4201834fe822169794f7 : RETRY_LIMIT in 36m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/57bf06052f8e421ab5c199628bbc0078 : RETRY_LIMIT in 37m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d24c124efc98498fb407407da32bf8cf : RETRY_LIMIT in 35m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c3df67f185542e9bcaeb59c081ca0ec : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":3},{"id":"37f042efe349deeb8684af63b1b4e3029537f97c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-27 20:31:15.000000000","message":"Patch Set 3:\n\nrechecck","accounts_in_message":[],"_revision_number":3},{"id":"20ad526df2b6a128601129531b9322bd16416982","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-28 08:15:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f2f377170013fc9fbe41b7b494e2bd7c3ea9e45a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 09:58:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/57e4e52ac4f543ef96935846d296bced : SUCCESS in 9m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cd5c9c9a8ee14c54a208358c1d7e2fac : SUCCESS in 1h 09m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/514b6a1ae2984e21b29f00cb575b080d : SUCCESS in 1h 37m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ca014041b67d4964a2ed49a87d916935 : SUCCESS in 1h 23m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67ad12b5e46c4168a5167193d09aa1c6 : SUCCESS in 1h 19m 26s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d4dc5e470c9f4094915ca948f16e6bce : FAILURE in 1h 30m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e0272b488f343219f56d1c161a2bfb2 : FAILURE in 1h 06m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bfc4669f436b46479e3618b3aa5856d8 : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a64ade14386b4a2da822ef8f539c2969 : FAILURE in 32m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14438b62432d412987abeef8a5c6a0ba : SUCCESS in 1h 02m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b10817b9b34ae6b576dd44fc52166c : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":3},{"id":"cd82c1a8350d2768989ed687d62de3a461c2def2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 09:58:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"800f09f62dc8839cde351228ee5b9f2322da9801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 13:03:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ada6705055844854a40a97be8798de4b : SUCCESS in 9m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/21e8458ef8ab49cb917008dc244eddf9 : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/487cbc86b7f2419d9ba0cd654d375b85 : SUCCESS in 1h 48m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b4b3f2e5ff2547eeb24893c94c68000d : FAILURE in 2h 42m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aefc4dc1c5474cd2b35339f453de49f3 : SUCCESS in 2h 02m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/616643b60d8b48c09c8c2ac7bd56276b : SUCCESS in 1h 20m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4cb1249d042434a9b03dbfc2046fad3 : SUCCESS in 1h 08m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb17ead357945be9dd4346c569e7c25 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"6ea5a2a74879b47272462987ef560b9e4dcef477","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-28 20:03:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"16088cda31f9678ca3a17c3c9457b71314680618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 22:02:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c9a8cfa906ba418e8211533756fb1657 : SUCCESS in 8m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/26dc23bab8664bd88b4cf28fbd3292a9 : SUCCESS in 1h 07m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a18946bbe27b419c97055534bc9ee60a : SUCCESS in 1h 50m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/727ca0dbe491424aa9e3c9fd2757c127 : SUCCESS in 1h 36m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c6b29bcf45441dc8c495e2e00914e53 : SUCCESS in 1h 14m 43s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/67f7019690ec46d9a162f10ca1dbb718 : FAILURE in 1h 32m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fcbf750d4864b5cafa2a4a69d5fe2f2 : FAILURE in 1h 08m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2df681aa672f425ab7c0d2a614d2112e : SUCCESS in 1h 30m 53s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0682aed3e5314fd4a89efa0f4e50048c : FAILURE in 31m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ee8600bc6b746ee802cb91fec182736 : SUCCESS in 1h 00m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e64e17d285e404c87d53f2fa6b2e935 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":3},{"id":"13842d07514845c0566213a3f2e46ab7e4015d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 22:03:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ccb770869f26ebfd63245b247bb9b77d146c88ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 23:49:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5ee8ff17e0c6e4bd396d061a6c572d8d9d5d3992","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 23:49:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f7a65686aa814ab49c2bb66a6ce09f7e : SUCCESS in 8m 37s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/254b166ac3e94eb18f09b72c9c018617 : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4a0715dcfb974eb68fa69ba10dd37157 : SUCCESS in 1h 40m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/94ccd4abbadc44de94967f8db69c7397 : SUCCESS in 1h 28m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5965b58f00bb47fa870c1dd8c13284d3 : SUCCESS in 1h 16m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ac543af39e16437587b1a5b2b4fd2ab2 : SUCCESS in 1h 17m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b3f1c2a583d4f2086287a3bdc2afaeb : SUCCESS in 1h 02m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde4514fba4344a7ab7939ba01216d92 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":3},{"id":"f876bab85c6685e5722a792a1cc80fe63aa65b22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 23:50:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ecda2ab8224285a2fdfa1d8090d0e7 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1993c6e4a7a0a4bca26e96a5018d00eae4cdaa36","revisions":{"598365d4beac9e211449492a2dba35d221ed41fb":{"kind":"REWORK","_number":1,"created":"2020-01-31 16:03:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/705247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/47/705247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/1"}}},"commit":{"parents":[{"commit":"1e6cc45d97dc92767fec9abb11c7d275f2f657d3","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1e6cc45d97dc92767fec9abb11c7d275f2f657d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-01-31 16:03:46.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-01-31 16:03:46.000000000","tz":120},"subject":"Use py2 shebang for centos distro installs","message":"Use py2 shebang for centos distro installs\n\nSince CentOS 7 do not have packaged swift library for py3, py2 shebang\nshould be used in python script so that they could import libraries\n\nChange-Id: If42feb0bb3ed0de211e5fba65c59d9ef50d199ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/598365d4beac9e211449492a2dba35d221ed41fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/598365d4beac9e211449492a2dba35d221ed41fb"}]},"branch":"refs/heads/master"},"5497e629726aad465e5226fa7e8d1f4d0cc0a3ae":{"kind":"REWORK","_number":2,"created":"2020-02-11 07:14:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/705247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/47/705247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/2"}}},"commit":{"parents":[{"commit":"1e6cc45d97dc92767fec9abb11c7d275f2f657d3","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1e6cc45d97dc92767fec9abb11c7d275f2f657d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-01-31 16:03:46.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-02-11 07:14:26.000000000","tz":0},"subject":"Use py2 shebang for centos distro installs","message":"Use py2 shebang for centos distro installs\n\nSince CentOS 7 do not have packaged swift library for py3, py2 shebang\nshould be used in python script so that they could import libraries\n\nChange-Id: If42feb0bb3ed0de211e5fba65c59d9ef50d199ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/5497e629726aad465e5226fa7e8d1f4d0cc0a3ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/5497e629726aad465e5226fa7e8d1f4d0cc0a3ae"}]},"branch":"refs/heads/master"},"1993c6e4a7a0a4bca26e96a5018d00eae4cdaa36":{"kind":"REWORK","_number":3,"created":"2020-02-11 11:30:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/705247/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/47/705247/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/47/705247/3"}}},"commit":{"parents":[{"commit":"1e6cc45d97dc92767fec9abb11c7d275f2f657d3","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1e6cc45d97dc92767fec9abb11c7d275f2f657d3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-01-31 16:03:46.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-02-11 11:30:06.000000000","tz":0},"subject":"Use py2 shebang for centos distro installs","message":"Use py2 shebang for centos distro installs\n\nSince CentOS 7 do not have packaged swift library for py3, py2 shebang\nshould be used in python script so that they could import libraries\n\nChange-Id: If42feb0bb3ed0de211e5fba65c59d9ef50d199ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1993c6e4a7a0a4bca26e96a5018d00eae4cdaa36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/1993c6e4a7a0a4bca26e96a5018d00eae4cdaa36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
