)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~718208","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Ie73811e36f1ea35f060cb5af4e1275e7e69d4179","project":"openstack/openstack-ansible-os_swift","branch":"master","hashtags":[],"change_id":"Ie73811e36f1ea35f060cb5af4e1275e7e69d4179","subject":"Revert deleted config section in Swift config files","status":"MERGED","created":"2020-04-07 20:12:17.000000000","updated":"2020-04-26 21:52:21.000000000","submitted":"2020-04-14 23:47:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718208-1586908057626-79213a0a","meta_rev_id":"137a5df45d4a828c91532216de2bdf87bbcd14ba","_number":718208,"virtual_id_number":718208,"owner":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 23:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-14 18:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-04-14 18:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-14 18:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-04-14 18:13:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 18:59:46.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-10 18:44:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-14 18:13:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-14 23:47:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65ddae4b456fed374b16b1b7c192be77e805a4ec","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-04-07 20:12:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f0e729c9922f99e2d06bf56f4ef3111da93173f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:18:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/90c37a8adb274fbc87be222e97af93a7 : SUCCESS in 8m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dab948cd58da4518a6c05b33c7431cc9 : SUCCESS in 1h 19m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/132a530ec424423aa74f52c22a8cd102 : SUCCESS in 1h 39m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ceb66a3901274c30bf8ea2f57097e95a : SUCCESS in 1h 43m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfd3249a28dc481294afbbfe89336376 : SUCCESS in 1h 17m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e56140e89ed34d97b7f1f3acb8ad194c : SUCCESS in 2h 48m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15ee0f96de7643bfb46a4029db40e440 : SUCCESS in 1h 57m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ddb5e4965ece49f39851f0d1567dda94 : SUCCESS in 1h 16m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/adaa55f68675448089ad1e412037ca89 : FAILURE in 39m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ae0230622ef473bb2b9f083d3734472 : SUCCESS in 1h 26m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08fc1802fb8843c3a4d50185a3f8f5e5 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"64b7733bd6985422b141a5b22d636e19fa861d24","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 11:21:52.000000000","message":"Patch Set 1:\n\nLGTM, but seems account-auditor and container-auditor are still not happy with configs though\n\nhttps://zuul.opendev.org/t/openstack/build/bfd3249a28dc481294afbbfe89336376/log/logs/host/swift-account-auditor.service.journal-21-41-40.log.txt#7833\nhttps://zuul.opendev.org/t/openstack/build/bfd3249a28dc481294afbbfe89336376/log/logs/host/swift-container-auditor.service.journal-21-41-40.log.txt","accounts_in_message":[],"_revision_number":1},{"id":"8ed6fa8a86d3f197cae724016fafc0b262366171","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 11:34:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"99855a43ffcd3a0d7862468fba6e6b5eb0dc28cd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-08 18:59:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e8b7d2111e92478d156f6ec47024494e8ae29bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:03:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a751498a621765d26403d03dfc0d19880d724706","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:04:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9aa261214a204b37b82bc04161f4b3ce : SUCCESS in 8m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ab5a1c72006746a6bc75e93dcadc63c7 : SUCCESS in 1h 26m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cfe8d60e004d44d093a5bbd9b306d2dc : SUCCESS in 1h 54m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a844681c6c91456a8945e298f7d66f7c : RETRY_LIMIT in 9m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba31e2121a3c4d7bb97fae75d498c040 : SUCCESS in 1h 45m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f955be8ccb344b9c838ca12dfa9c93a1 : SUCCESS in 1h 25m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94bc8d5cb4ce4698a7cb461e85b7cefa : SUCCESS in 1h 14m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06104a9cad81495b8a18baf804e135fb : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3eb9a68b334bf6d08573b4b92c2635c9b1d81f30","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-09 11:52:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbb20c3ee2192dace3074f8f339034b50dc7025b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:01:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2a779a71111a47909f044f1a2afdc3b7 : SUCCESS in 11m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b4d99f0af5b54a749dc7d65477040dd1 : SUCCESS in 1h 24m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/35f1f933daf440b3be572190f42f2350 : SUCCESS in 1h 34m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f7df5b2c0cf54f059135050ec9055dad : SUCCESS in 1h 59m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58050abceb02418885194cc3485db59f : SUCCESS in 1h 24m 47s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dc9012e445794615878a3e7150f1639b : SUCCESS in 2h 58m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08a16b9aff044bf4925aed2aa27b215c : SUCCESS in 2h 05m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2cf478ca4d6b4553ba0efcb08abdc92d : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/54a563bf81df4d80b577316b16223752 : FAILURE in 37m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65eea15fa0e44c42b9c55ac0a359270a : SUCCESS in 1h 04m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2946348ec4f429c95c1ca242edba737 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"05dda7898ab88a76ae28c90ae3b92001c2c214f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:01:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a23eee77d7a67a6815e9e3755d9b918dd47fe9b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 17:22:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ad76aab597f54ee3a516ac00240ef92b : SUCCESS in 1h 21m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d8dbeefde5fd4863a71880e5b8bf0400 : SUCCESS in 2h 12m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/df4d741053f14a10a5899bc0c1813c23 : SUCCESS in 1h 43m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc17597f07f441968dacd950175581bb : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9376653bce294fb6817d05062027dc44 : SUCCESS in 1h 37m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb4b55ece52d451ea714a076ac7afd10 : FAILURE in 1h 05m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f634af5a71ef4a28acfb4de1f0642de0 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3f79fe50e1a33865f58114452347bbe3d2ba0a70","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-09 17:49:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"980e18bfbebd0b191f6f003c016eb16bc4456568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:31:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7df8e967286c40949c8b859ce7c40a7c : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fd4fec9670f84056bf6670063bc226bd : SUCCESS in 1h 49m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/982d9f7f8c1942dbadfa5d344a7b0c0e : SUCCESS in 1h 36m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23d58fbc13834dada4d1ca946bb57d85 : SUCCESS in 1h 24m 14s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/08f7f9fbe12949a8acfd11f4b9ae3c11 : SUCCESS in 2h 36m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0225ba5575144b21ad00411bf94e0eeb : SUCCESS in 1h 54m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5fbe2840ffde48c999fcdfec46bab885 : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d0d56764e5a84f9fa8e9ae289e075043 : FAILURE in 40m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5698ca4f4e94a6e8778c52601d57ab9 : FAILURE in 47m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1238a2b79eb437e97f7f269f6ba02ab : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"452841842ea1f66f01dc94a98a61659f3a29afa9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-10 07:01:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b698efac85456c4d45d3ee7a5516785a604043a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 13:28:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d0331898d4644310bc562d9874a7877c : SUCCESS in 1h 09m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6bb057e272984368bf86e792e276e01a : SUCCESS in 1h 52m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7d4e3a33482f4850931c02c35946fd94 : FAILURE in 23m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59247ba1ddd84db59f15fa7779a53851 : SUCCESS in 1h 39m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5b66af5fcef54e01ad6a67be391f65da : SUCCESS in 3h 01m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/785a0643f9b74fb2bbed434256731ace : SUCCESS in 2h 12m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/54ed449cfee840ea984d9f9c75741c73 : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/33897c9c1b2c4c67bde5bb2e3928f0f9 : FAILURE in 34m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df607b280e334cfbb3e2b18c1c6bb31b : FAILURE in 43m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5746be6a39ec49b887aef8a9243f24da : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d0e936677652f46af3eab6f38829a26d3fe43656","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-10 17:02:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13a6d685a649981f6b1250c08507290da69ca6cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 18:44:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"021e29ca56d1e57da3c21728c3eaf51946738800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 20:07:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/21141126fd4b44cf9bd68a35b96b9a9b : SUCCESS in 5m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7a685067593f4a2f8ab5be2eaf121f7d : SUCCESS in 1h 09m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3878ed0eb18e4d179b082ebc8a47100a : SUCCESS in 1h 33m 28s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8975511f2ff94f2c94c1571977dc7a7a : SUCCESS in 1h 27m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56cef11036954ea186e8cded1eec2121 : SUCCESS in 1h 13m 12s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a98d54712543436a954c1a77725a4c01 : SUCCESS in 2h 57m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0422737fe3446ae9af8c7879808b0a5 : SUCCESS in 2h 03m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f0718f4296204aaba4cf37a767c671d9 : SUCCESS in 1h 19m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1483e1e52e064728a1fa2407e4dd04e2 : FAILURE in 33m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9eff4c6f97754398b9ac46a6209ce9ed : FAILURE in 45m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99619d876374689b72ef5a7ceb89e80 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c23b26d2ff95e666c33e7f97432213bc900162eb","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-04-11 23:41:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a01330747a19eae396962a520915ef35d6f0678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 01:30:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/28b376aae9084656894518c944f977bc : SUCCESS in 10m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f3f2d0dd151f4373a00ca34e5306bd9e : SUCCESS in 1h 14m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d8e3ba8e99b4574ba42f58fbcfb88c2 : SUCCESS in 1h 38m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/acb2b031430b4ca882528e66b293963c : SUCCESS in 1h 27m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3dc7c8020c82418d890352c4d7985e8d : SUCCESS in 1h 19m 33s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ac9b8feef33944b0b9888a1ae161e43a : FAILURE in 1h 27m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8793377234754eb499cb64498786c51d : SUCCESS in 1h 46m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e720d8e36bf64b2794ef1193d8e5ac84 : SUCCESS in 1h 07m 10s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0d713698a9474326be468dd34fb783e6 : FAILURE in 28m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e36301659a4e4c49895ac45bb71a457c : FAILURE in 46m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4c6cb4234240ff9ba44b05efed267f : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e3d88234d8bd8c3bc3e6737c05911ea81409518d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-13 17:08:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d0e262fab7e41f12f449f13d44cefb81f03fd469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 19:40:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3dc3fe987ff54b669299860be8fcef32 : SUCCESS in 7m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c241942ca3924377b2744f12f0761048 : SUCCESS in 1h 14m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e0c01817e6f64f24938de1e38e4edd58 : SUCCESS in 1h 35m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6c1332e5fe43426c95f744dbce328dff : SUCCESS in 1h 38m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b7347d6d5f74eb782dc44eccaa6c0cf : SUCCESS in 1h 16m 10s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b8022bc69ebe4b509b170daaa1f17c18 : SUCCESS in 2h 25m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15dc3c89612b4230b7cf3a794cd5704e : SUCCESS in 2h 16m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fd765f9d0a2845039668fe14ae9739fa : SUCCESS in 1h 31m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a833d71cf9ae4a05b590699d9e519acc : FAILURE in 33m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3749fe8baa714ba39c71a99e51695c77 : FAILURE in 48m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30130058288d49c888e9bceb0c6a4be3 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"448c0faa3ea9f8d899c667dfb55f1e08f8355d8a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-13 22:00:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76fee69cc99774c414a13ef43d40a80fa3accc63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 00:48:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/769de3e0d06849358c71194e737fd2b5 : SUCCESS in 6m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c62e99bc1fc747e19a3b2fb8208fe23d : SUCCESS in 1h 10m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d354ae5a1c3452eb4dfac0a1ed79e31 : SUCCESS in 1h 49m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/282158f1bf784b3e9b01643604c7420f : SUCCESS in 1h 32m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b5bb9c3619a4c829e5240f21725e9c3 : SUCCESS in 1h 12m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b2ed0d8bbfac4de2996171324235c4b8 : SUCCESS in 2h 44m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67785d4ba64e493a92d9aa00c8c222e0 : SUCCESS in 1h 57m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d094eff7c1484274a886b3e350c10d97 : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aff90cb6081f4e08a756bd4c6640b5e2 : FAILURE in 34m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bee9f41ea0644ac9d2959918bf163dc : FAILURE in 44m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a67ae8972be34ce7a1849d5c9da1072f : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"15bef0639942de26410cb18c8fc6c7e282beb62b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-14 14:00:44.000000000","message":"Patch Set 1:\n\ninvestigating/fixing ubuntu distro deployments...","accounts_in_message":[],"_revision_number":1},{"id":"243c275a885ae62b3a2c4737019ceb7556362714","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-14 18:13:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5b284f4f9af2f98b77c94428a4bd7a9f3ef1addc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-14 18:13:40.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f076b6865f38283b0c4759955855caa40868687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 21:02:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dcb310c92cb84e3fa2e9670bb4e54761 : SUCCESS in 11m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d3e838e2ad46431abfd490a65456b4ab : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cb26260bffe146d59511c394fb2bd710 : SUCCESS in 1h 35m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fa4d6d30e96344388f39b1494d2899d4 : SUCCESS in 1h 31m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c5f00be6cf8483bb79ccaf8e77c30fc : SUCCESS in 1h 30m 46s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7091b1885e0e4d0f8c8781c154c271aa : SUCCESS in 2h 38m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/304faef87bab48fe85aacb6a7697262b : SUCCESS in 2h 36m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8161043827014263958e282174bfe03a : SUCCESS in 1h 22m 12s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4eef6fbbd595499cb619ff6c60064f13 : FAILURE in 38m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6ea9dcab2f145839cbb791ce708ffd1 : SUCCESS in 1h 04m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05b0891a0bc42f9b2a0ce5e774379f3 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":2},{"id":"5151a14ca0d6b7670e549a8ed02d4d0ebe39dc21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 21:35:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de6455053523ea75be41e873be0def6fca4727b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 23:47:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/06e92d7566354420b8c7311dedbc6767 : SUCCESS in 9m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6d964cdd33194a08b3759d039b532e52 : SUCCESS in 1h 12m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8a3628da36f24ac6b1061941078aa0c6 : SUCCESS in 2h 05m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/56e6e30101d646eaa667e43db815b5c8 : SUCCESS in 1h 40m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f193d04708f4c25b689d2aedd5825d3 : SUCCESS in 1h 34m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ce0fdaff26614f80a76e17a13ef5a901 : SUCCESS in 1h 35m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed9d539ebd714d3c8c3fb5ff54fc37eb : SUCCESS in 1h 43m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e666ded37551436fbb112fe6edeb504f : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d4cdd307bbb2196ae9f2dcfd270d5fff5a93c428","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 23:47:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6496a2f4ac7f12bd7aefd74134a8338510dfd4fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 23:50:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6202ba8e546430db88a892dfc2a2831 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":2},{"id":"137a5df45d4a828c91532216de2bdf87bbcd14ba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-26 21:52:21.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 67cbc530bbb9bf4fd95542683b1179bf068bcb5d","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"91f26a98cf067a963ae7dc4809ce1438b70e0279","revisions":{"194d01948a3b2f15747493a5811bc68785edd8e6":{"kind":"REWORK","_number":1,"created":"2020-04-07 20:12:17.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/08/718208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/08/718208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/08/718208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/08/718208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/08/718208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/08/718208/1"}}},"commit":{"parents":[{"commit":"677ab44170bbf60fc422fe843073389906d3fe45","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/677ab44170bbf60fc422fe843073389906d3fe45"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2020-04-07 20:07:20.000000000","tz":-240},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2020-04-07 20:12:16.000000000","tz":-240},"subject":"Revert deleted config section in Swift config files","message":"Revert deleted config section in Swift config files\n\nA recent commit [1] removed config sections related to logging that\ncauses various Swift services to fail to start. This patch reverts the\nrespective config sections. These may be removed at a later date.\n\n[1] https://review.opendev.org/#/c/671711/\n\nChange-Id: Ie73811e36f1ea35f060cb5af4e1275e7e69d4179\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/194d01948a3b2f15747493a5811bc68785edd8e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/194d01948a3b2f15747493a5811bc68785edd8e6"}]},"branch":"refs/heads/master"},"91f26a98cf067a963ae7dc4809ce1438b70e0279":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-14 18:13:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/718208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/08/718208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/08/718208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/08/718208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/08/718208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/08/718208/2"}}},"commit":{"parents":[{"commit":"34419fd92b4737bdde2de2252f7f94179de0e903","subject":"Fix distro installs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/34419fd92b4737bdde2de2252f7f94179de0e903"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2020-04-07 20:07:20.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-14 18:13:36.000000000","tz":0},"subject":"Revert deleted config section in Swift config files","message":"Revert deleted config section in Swift config files\n\nA recent commit [1] removed config sections related to logging that\ncauses various Swift services to fail to start. This patch reverts the\nrespective config sections. These may be removed at a later date.\n\n[1] https://review.opendev.org/#/c/671711/\n\nChange-Id: Ie73811e36f1ea35f060cb5af4e1275e7e69d4179\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/91f26a98cf067a963ae7dc4809ce1438b70e0279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/91f26a98cf067a963ae7dc4809ce1438b70e0279"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
