)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~718524","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"osa/swift_replicator","hashtags":[],"change_id":"I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9","subject":"Fix replicator services","status":"MERGED","created":"2020-04-08 17:13:13.000000000","updated":"2020-04-26 21:53:00.000000000","submitted":"2020-04-16 05:36:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718524-1587015406905-0447f3fa","meta_rev_id":"1744fa8f57106b44907214ffc770e77969748c7e","_number":718524,"virtual_id_number":718524,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 05:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"recommended":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-15 07:00:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-04-15 07:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-15 07:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":1,"date":"2020-04-15 07:00:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":1,"date":"2020-04-15 07:00:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 19:51:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-04-13 22:03:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-14 17:40:32.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-14 21:35:43.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-04-16 05:36:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b13f57fb1dedaeb4e7363a4eeee85750d0fb1b8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 17:13:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"081ed0947e6cd468b358bd708292b619c07c3339","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 17:13:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1e68e77228a88310445af0f816b240d07bbc5026","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-04-08 19:25:39.000000000","message":"Patch Set 2:\n\nI am not really sure this is necessary. My recent AIO deployment\u0027s Swift-* services started OK when the bidding config options were replaced.\n\nSee:\nhttp://paste.openstack.org/show/791826/\nhttp://paste.openstack.org/show/791825/","accounts_in_message":[],"_revision_number":2},{"id":"4a8a81fb12477308d94dd03bf3fa79e606349eb7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 19:31:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"37f8504917bd295f2726af7d7a5ebd742e8105f8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 19:40:32.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"08e91a7163d82c4624412940139bc14562c24974","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 20:05:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"57d79fb7e71180a53ab53811159933b002185d04","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 20:07:29.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"36c24754a7e79d1c893d340f8d92968c6a0c8c2a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-08 20:26:28.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nSorry, depending on config, Swift just needs a *lot* of services...","accounts_in_message":[],"_revision_number":5},{"id":"5a8b632b651a01e065e4268ed8180256f33b1afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:23:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8dd6a30c7bfc400e9aa59520750d69cd : SUCCESS in 10m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e261b4f4933442a3a2440e068f0a8e8c : SUCCESS in 1h 09m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c717a852c96b49918230c50cda23996b : SUCCESS in 1h 32m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/78470396ced0451a8c400c713122cfca : RETRY_LIMIT in 20m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2d0af620891439687241db405281dd7 : SUCCESS in 1h 22m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c169df8523ba42d0b79c5f33ae98ed25 : SUCCESS in 2h 50m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d800ddf9b6db4ef180f1781cc88d3898 : SUCCESS in 2h 05m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8c1b29470d114caba41e1a913f41a5f1 : SUCCESS in 1h 15m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d436291adf5a445bbe5108ead7bc9d23 : RETRY_LIMIT in 11m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe806f1260db4291aec0ce4542f5e26a : SUCCESS in 1h 05m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a60e9830fa004b9491866343b498db05 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":5},{"id":"6592b7af471f887824522683787a48ff48d1febf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-09 11:49:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a7282568d58210d60f3cdbffb795ae03d6d935a1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-09 11:54:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"43239c9478617c1a9edeae33c42018fe1c27b4ce","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-09 11:54:54.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"289c61a7405ed88372013590c7c1a544d8d36267","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-09 11:55:42.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"fdb68b6e398c82310899e745454e8671aa775f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:41:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/748bb75e7c1143caa11c0f599e13478c : SUCCESS in 9m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1fdfb3143fad4ea3acd395636fe3d589 : SUCCESS in 1h 09m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0c7ff2e63a0d4201941639d5040edc76 : SUCCESS in 1h 48m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/db6e834714ae4c1b95679e3495ebde24 : SUCCESS in 1h 31m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ffa556209bf42d4b99442ae014edc51 : SUCCESS in 1h 18m 21s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5c34468da5d84031aafd714b73fb6aea : SUCCESS in 2h 37m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/255ed04992494d0fa176d32897d7e788 : SUCCESS in 2h 02m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/85e1535a28da4ca3b34b145094f293e0 : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cd7ae32658f045b881c03c84f0145117 : FAILURE in 36m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d3fecbfad624f1da6f339e3870a6033 : SUCCESS in 1h 29m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36cca15c9f4740ba90a34d645724123f : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":8},{"id":"95a5de6462454d960eb51bf7c1f24d05b1c6ed90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-10 19:51:40.000000000","message":"Patch Set 8: Code-Review+1\n\nLooks sane to me!","accounts_in_message":[],"_revision_number":8},{"id":"bfee8099c44fbb8cffbfcc68a850f84a187b41e7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-13 22:03:31.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d31a4f6ced0ea865fd816196a072ce6b4aa8acd0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-14 17:40:32.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"90a15ab96bc7460786ccf27d18cb88e333ac2fd8","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-04-14 21:35:43.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"7ee31bf8a3ce24c492bf330727e33ab534278037","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-14 23:54:35.000000000","message":"Patch Set 8:\n\ndoes this need a recheck?  I don\u0027t see it in the queue","accounts_in_message":[],"_revision_number":8},{"id":"90d78c2385752d759e4ca072be1117afeb794bb2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-15 04:53:55.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"02cf4355203fdb47a4025dfda61a239224a1d7b3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-15 06:15:19.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"8782db09b339c06a14107f30c98a1b8a51500e41","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-15 07:00:12.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"52119f8738ebb8d1c0af752ce2a897fde98dcc28","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-15 07:00:16.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"453f9a5a4d5969f0c1f7e97a0f12fe53b26e812f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:19:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e5dbb8a2824942c38263836fd4f0ef0c : SUCCESS in 9m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/205544a9bd8d4376bfe98227be48ef31 : SUCCESS in 1h 22m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bf039f6e065a4f098cff2db1ce4c9362 : SUCCESS in 1h 52m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fcdd0516f9af43adb8046cf45596135d : RETRY_LIMIT in 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da029f72527c48889bf64691cab76d5b : SUCCESS in 1h 20m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5503aacc7f9a426589df9a79808aab9a : TIMED_OUT in 3h 04m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fb69c226b574919bfdd259442fd8f08 : SUCCESS in 2h 01m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5cfbba298c9a45498e267703081a98c6 : SUCCESS in 1h 15m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7b1c8cec293e4d0a8ae19f23ce4e7432 : RETRY_LIMIT in 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e71dc2082e04ce89859d4e9e7d84931 : SUCCESS in 1h 08m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/552878363a974fb988d950d25a4136f0 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":9},{"id":"bc85c17e7ceadcf7cd21cd6b8eb9041e83b2a2a4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-15 11:16:14.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"4c9c556a70c629272997753583d0fe145ef9b3da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 14:01:26.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4f5c1ecee50e4f3facb4aa9f203c3f00 : SUCCESS in 9m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/683f3ffb0fc84bea9fd396dc2e69d2f4 : SUCCESS in 1h 11m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d0e5b2b1af804f5d9f809bb233d883e6 : SUCCESS in 1h 46m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/04bfd1665db448d99eb63b6e8c61e93b : RETRY_LIMIT in 9m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7861486d67c0419699363a7a91ae04af : SUCCESS in 1h 19m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/41d39556e7e14b91994814c95a9e2abd : SUCCESS in 2h 39m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72c784365a6348349051cc1ee2f7f716 : SUCCESS in 2h 06m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4433e0ecc59342f7a8e644f864d99bb2 : SUCCESS in 1h 19m 40s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/df60e93c46a8432abc2f12e742fa3ef6 : RETRY_LIMIT in 7m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d1542e08c334777be117c7a2b2f433b : SUCCESS in 1h 00m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f99c5f40c6d34701a593ecd135504bc7 : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":9},{"id":"d24086603f46d2f7a48e7eeeedd53203628d8e6b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-15 14:35:07.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"43e823c848d657873603f5ead5cacfb42db9a031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 17:25:47.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5cf4f5f6aa4e4b84ac206993857e7d0c : SUCCESS in 9m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9adcedc23f8340e89f00ca5ecb96b2b1 : SUCCESS in 1h 15m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cd5595d6095342269a9e550c39907e2d : SUCCESS in 1h 48m 10s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cefbce9eda3a49688f2c06efbe3b3cc4 : RETRY_LIMIT in 15m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99ba75d60da249f88bee0fb7be616ecf : SUCCESS in 1h 19m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/350148fe8dba4d15b8e31704bfa427d9 : SUCCESS in 2h 43m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cac5022f92249bea3fe7dcd15001ae6 : SUCCESS in 2h 28m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/387fc800884e42f8af83b866229a247c : SUCCESS in 1h 46m 12s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2f9d3ef37597422089a6ef746b3942e4 : RETRY_LIMIT in 11m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e479f3aa4e2455a8ff864f51cf892b0 : SUCCESS in 1h 05m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25df4987594435abee43593c98f171e : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":9},{"id":"f02616411b317b45f0a47ef84d4fd746fecbb945","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-15 17:54:31.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"61d83c8da113c905448a53166e7886f2a7589331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 21:01:14.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9f517b79aa2e4f129194654b4ae42afa : SUCCESS in 7m 17s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/30edc5a732704cc384a064f7954c6bd5 : SUCCESS in 1h 11m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/adef7bed40c04c488a02b9749cf740aa : SUCCESS in 1h 44m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ef71aa53b8048558db5df99a93c615e : SUCCESS in 1h 20m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/021e94210dfe44fb8fd237631dc52653 : SUCCESS in 2h 59m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abde2d014f7548d1aef43aae713b0661 : SUCCESS in 2h 13m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e4bff2e03107456aaff331bb8651c4ea : SUCCESS in 1h 17m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5a19d0762ed47b9a2f0f3d19d47bf5c : SUCCESS in 1h 00m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9066b118634284a1792e9217523ed4 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":9},{"id":"09dfc3ea51b2c227f3fca50192ebfc4ba2e73e32","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-15 21:06:42.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"42635dc451d0df16203e8abc3af50414a824cb2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 00:08:24.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9df336257cb549e6b9041f4cbc545fdc : SUCCESS in 8m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/01d0cbb27c194d18bc18cdd3539a2a3f : FAILURE in 1h 30m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4d7fdb5681f9428daf99a32ec40bcdcb : SUCCESS in 1h 54m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/22dc4a364fe44ff9a409ba531d081f38 : SUCCESS in 1h 30m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d1b84863afc4c688fe3a4939ffe203e : SUCCESS in 1h 19m 50s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a3e2a12b2d97429bbce3c85992a76459 : SUCCESS in 2h 55m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e50daf2754e4e518306065be1f32b46 : SUCCESS in 2h 12m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/23c0d008ba2b477092b094c96f27ea69 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b6c30ce77b7b4572a93778119025b306 : FAILURE in 36m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8585cab79a2f404a9761875a41b25247 : SUCCESS in 1h 02m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c7aafe045d4f07b2ccfc49eedf5713 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":9},{"id":"e269b4e2bcb7a8e3717a27ab0567eaf5a265dcce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-16 00:51:55.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"ae620e2e4e028dc5b9e4a3b42246cf03d49b1a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 03:55:45.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bd9092205d764f45a94b754f09407c86 : SUCCESS in 6m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e1dc73be94af4c4b9704a0d73255e858 : SUCCESS in 1h 07m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f2aa2f8c9dd24768a48529a5acf1db5c : SUCCESS in 1h 41m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/89214ebe74364d6da2d12509f6a78f0c : SUCCESS in 1h 28m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c977f4ccf7d04b2cb59a1b24907a8971 : SUCCESS in 1h 14m 54s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/74e9e5eb805b4207bb0334480cc636ca : SUCCESS in 2h 56m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b8cf7db4b004d1b91c9a5a77d92f473 : SUCCESS in 2h 06m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3e88c3b2ecdf4351a7d3a18555ce20ac : SUCCESS in 1h 13m 44s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d6471e335b3d4557b97e67545892b5c9 : FAILURE in 31m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afde7bf9a31641b9adcfad607a50a063 : SUCCESS in 1h 04m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22a14a274534f80a1bd3f6f39dd81bf : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":9},{"id":"02fb28f3fa902ba4178d3c9002d8b0a1c0ac910f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 03:55:57.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"562cf3db7594c71908592557b2db92534753cc39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 05:36:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"672d88556ed1c13e142b2b58e66e8e555c00b205","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 05:36:46.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a0432dad1bd3476c83ff82cf07c63e93 : SUCCESS in 9m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4c4e11a6a28d42a6a75d22b62222bf82 : SUCCESS in 1h 18m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bb24c3a6872d4128b29ef5cc43bbe800 : SUCCESS in 1h 35m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b937a83cb4854ca7b9594583114253b3 : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2eac7ec864149bd8b961ba9d35d0917 : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/28b74ab2991b4994906fa44dcbcd4950 : SUCCESS in 1h 35m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33c411b5fee841919bbbda58142aed04 : SUCCESS in 1h 21m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b91d3d1fc52646998d8c78ef7e19a87f : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":9},{"id":"04215455a52aa11be8bd1f72e9defe947e0017f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 05:38:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2dff0fc11e5494787db09fa02ff3ac4 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":9},{"id":"1a1b056214a406c4db37abe0b639d00cd7e0ca03","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-26 21:53:00.000000000","message":"Patch Set 9: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 8a83c9183edb69e1809009720b2e1f2eed3a4a93","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"be9ad06226d74cfcb796834d27ca641902329fcf","revisions":{"6308a9d656c3659a745ffa72647e030a854f6092":{"kind":"REWORK","_number":1,"created":"2020-04-08 17:13:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/718524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/718524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/1"}}},"commit":{"parents":[{"commit":"677ab44170bbf60fc422fe843073389906d3fe45","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/677ab44170bbf60fc422fe843073389906d3fe45"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:13:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:13:03.000000000","tz":180},"subject":"Fix replicator services","message":"Fix replicator services\n\nCurrent services were infinitelly spawning replicator process\nwhich is not designed to be runned as service\nbut just as an ad-hoc script.\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/6308a9d656c3659a745ffa72647e030a854f6092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/6308a9d656c3659a745ffa72647e030a854f6092"}]},"branch":"refs/heads/master"},"2fb4abfb8167cdef539a81c21b2a171fb79e87c5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-08 17:13:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/718524/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/718524/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/2"}}},"commit":{"parents":[{"commit":"8b1a0f5c0489f86961263c32ca2109a2091d7621","subject":"Revert \"Revert \"Update fallocate_reserve to work with py3\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/8b1a0f5c0489f86961263c32ca2109a2091d7621"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:13:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-08 17:13:31.000000000","tz":0},"subject":"Fix replicator services","message":"Fix replicator services\n\nCurrent services were infinitelly spawning replicator process\nwhich is not designed to be runned as service\nbut just as an ad-hoc script.\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2fb4abfb8167cdef539a81c21b2a171fb79e87c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2fb4abfb8167cdef539a81c21b2a171fb79e87c5"}]},"branch":"refs/heads/master"},"412b04dfc99517c20a96c2fb5dd55de548dec852":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-08 19:31:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/718524/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/718524/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/3"}}},"commit":{"parents":[{"commit":"194d01948a3b2f15747493a5811bc68785edd8e6","subject":"Revert deleted config section in Swift config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/194d01948a3b2f15747493a5811bc68785edd8e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:13:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-08 19:31:03.000000000","tz":0},"subject":"Fix replicator services","message":"Fix replicator services\n\nCurrent services were infinitelly spawning replicator process\nwhich is not designed to be runned as service\nbut just as an ad-hoc script.\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/412b04dfc99517c20a96c2fb5dd55de548dec852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/412b04dfc99517c20a96c2fb5dd55de548dec852"}]},"branch":"refs/heads/master"},"db1e196d6b5e29cbff303794e007002906df45da":{"kind":"REWORK","_number":4,"created":"2020-04-08 20:05:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/718524/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/718524/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/4"}}},"commit":{"parents":[{"commit":"677ab44170bbf60fc422fe843073389906d3fe45","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/677ab44170bbf60fc422fe843073389906d3fe45"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 20:05:31.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 20:05:31.000000000","tz":180},"subject":"Fix replicator server services","message":"Fix replicator server services\n\nIn case of enabled swift_dedicated_replication we had regular swift-*-server\nas well as swift-*-server-replicator which seems pretty unnecessary.\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/db1e196d6b5e29cbff303794e007002906df45da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/db1e196d6b5e29cbff303794e007002906df45da"}]},"branch":"refs/heads/master"},"610276df9acdb4ba52767f525638745069238f96":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-08 20:07:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/718524/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/718524/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/5"}}},"commit":{"parents":[{"commit":"194d01948a3b2f15747493a5811bc68785edd8e6","subject":"Revert deleted config section in Swift config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/194d01948a3b2f15747493a5811bc68785edd8e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 20:05:31.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-08 20:07:29.000000000","tz":0},"subject":"Fix replicator server services","message":"Fix replicator server services\n\nIn case of enabled swift_dedicated_replication we had regular swift-*-server\nas well as swift-*-server-replicator which seems pretty unnecessary.\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/610276df9acdb4ba52767f525638745069238f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/610276df9acdb4ba52767f525638745069238f96"}]},"branch":"refs/heads/master"},"2d4bf163b9d2d64290dc54be09ea60b1e22ad907":{"kind":"REWORK","_number":6,"created":"2020-04-09 11:54:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/718524/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/718524/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/6"}}},"commit":{"parents":[{"commit":"677ab44170bbf60fc422fe843073389906d3fe45","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/677ab44170bbf60fc422fe843073389906d3fe45"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:13:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-09 11:54:11.000000000","tz":180},"subject":"Fix replicator services","message":"Fix replicator services\n\nCurrent services were infinitelly spawning replicator process\nwhich is not designed to be runned as service\nbut just as an ad-hoc script.\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2d4bf163b9d2d64290dc54be09ea60b1e22ad907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2d4bf163b9d2d64290dc54be09ea60b1e22ad907"}]},"branch":"refs/heads/master"},"c7af366f25334e47696be36b322c7632eeec7be8":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-04-09 11:54:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/718524/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/718524/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/7"}}},"commit":{"parents":[{"commit":"194d01948a3b2f15747493a5811bc68785edd8e6","subject":"Revert deleted config section in Swift config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/194d01948a3b2f15747493a5811bc68785edd8e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:13:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-09 11:54:54.000000000","tz":0},"subject":"Fix replicator services","message":"Fix replicator services\n\nCurrent services were infinitelly spawning replicator process\nwhich is not designed to be runned as service\nbut just as an ad-hoc script.\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c7af366f25334e47696be36b322c7632eeec7be8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c7af366f25334e47696be36b322c7632eeec7be8"}]},"branch":"refs/heads/master"},"5457c7931567347988cb6b79e348d237125ae7a1":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-04-09 11:55:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/718524/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/718524/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/8"}}},"commit":{"parents":[{"commit":"194d01948a3b2f15747493a5811bc68785edd8e6","subject":"Revert deleted config section in Swift config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/194d01948a3b2f15747493a5811bc68785edd8e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:13:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-09 11:55:39.000000000","tz":0},"subject":"Fix replicator services","message":"Fix replicator services\n\nCurrent services were infinitelly spawning replicator process\ninstead running a server with another config file\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/5457c7931567347988cb6b79e348d237125ae7a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/5457c7931567347988cb6b79e348d237125ae7a1"}]},"branch":"refs/heads/master"},"be9ad06226d74cfcb796834d27ca641902329fcf":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-04-15 07:00:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/718524/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/718524/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/718524/9"}}},"commit":{"parents":[{"commit":"8bceafccddcf8538d12813bc6cc3f84d56fe1721","subject":"Merge \"Revert deleted config section in Swift config files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/8bceafccddcf8538d12813bc6cc3f84d56fe1721"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:13:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-15 07:00:12.000000000","tz":0},"subject":"Fix replicator services","message":"Fix replicator services\n\nCurrent services were infinitelly spawning replicator process\ninstead running a server with another config file\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/be9ad06226d74cfcb796834d27ca641902329fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/be9ad06226d74cfcb796834d27ca641902329fcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
