)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~723273","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Ftrain~Ie73811e36f1ea35f060cb5af4e1275e7e69d4179","project":"openstack/openstack-ansible-os_swift","branch":"stable/train","hashtags":[],"change_id":"Ie73811e36f1ea35f060cb5af4e1275e7e69d4179","subject":"Revert deleted config section in Swift config files","status":"MERGED","created":"2020-04-26 21:52:21.000000000","updated":"2020-05-11 22:39:15.000000000","submitted":"2020-05-11 22:37:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723273-1589236665746-eed6f917","meta_rev_id":"319edb0425b5dae047a19aa58f65987d568c8155","_number":723273,"virtual_id_number":723273,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"date":"2020-05-07 15:27:23.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 22:37:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-05-07 11:25:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-07 07:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-05-07 11:25:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"date":"2020-04-26 21:52:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 21:52:21.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-05-07 07:32:49.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-07 11:25:24.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-05-11 22:37:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5def0fa7f9a026dde215ca479f080fdd5b5a4a6d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-26 21:52:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f86b0b566ea10ff9448e3038fe765194114be24e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 00:19:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d231b5cb54a645e79f5dacd9cc4ebab0 : SUCCESS in 6m 26s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1ff3402504cd420f979494bf0ac9d7a8 : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/90f13e945f8c419dad7d0b5331e3ef9e : FAILURE in 1h 07m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0fb2a1c70d7f412e8fc22ca953621463 : FAILURE in 1h 22m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/518a71b917bc4f4a90dbdb2604a39e0a : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa5f081fdd4948008ab5a12614634f52 : SUCCESS in 1h 13m 54s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1228e3c8124f4d9287d66a56c7014e38 : FAILURE in 2h 24m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b526cb9aab3041e4b2db156cd1926a07 : FAILURE in 1h 07m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1534baa011cb41b8a2bdc9b3d5b3bb7a : SUCCESS in 1h 09m 52s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/db8ba481ea5748c4a427cec0a3b8c8d9 : FAILURE in 28m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9bcdfda3a9d47d4a4af0781b8c0f23e : SUCCESS in 1h 03m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fb567709484ee798cabe077ab6f435 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ce193bfc36d8928c38f9f435e8cd8d6782b042f1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-29 06:19:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e5b337100e04f528a0941cafef61064ca24cb268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 08:47:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e31d2c08e4f4463094b6424c1e3fd77f : SUCCESS in 9m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c5c4499be01f401e85db917ad0968cc5 : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8c6c20e6194a4908ac3636e00dd704c6 : SUCCESS in 1h 31m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/627c46f480304ca2b613ad524168e903 : SUCCESS in 2h 02m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9d34a4bc978049588349aba23edef998 : SUCCESS in 1h 36m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f44b954bbdc94a24843593b760788453 : SUCCESS in 1h 21m 10s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f664909bbf244d6693e619ef762730de : FAILURE in 2h 21m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca8d6a53a77847128ecaf38138b32679 : FAILURE in 1h 12m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7786f477011b41d6bf477e4db39a9eed : SUCCESS in 1h 11m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aab8325599a0404d953cd2de0b193d92 : FAILURE in 33m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab6237791fac4edfbfbe809ccdb0dcc0 : SUCCESS in 1h 01m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8f3e3707d74508994a9fd43f9342e7 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0042f0438f2fefcf67e7d808f612640550180c43","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-29 21:01:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"deb65538bfba19e9b3d41a9002d53e6f523f64d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 22:39:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cba479d78ea54b68b692f3023f97b722 : SUCCESS in 7m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7804e9ee1c2642a4a526498677b0698e : SUCCESS in 1h 09m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/4178029adde1483387c0fee75926079e : SUCCESS in 1h 22m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fd905e6a96a64ce19d4d76bc00236fd9 : FAILURE in 21m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9e313ff5ec1b4c39a8b2d9170478d321 : SUCCESS in 1h 32m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d07036b5e6a347a78919a8c2175188d5 : SUCCESS in 1h 15m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1c3f62387c7a42b3916e9286bebaf112 : FAILURE in 21m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b164b4e03e0442d892c5ee1aee7639b : FAILURE in 1h 03m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b2c5cc12fc52427b89a6947622c8f729 : SUCCESS in 1h 16m 31s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/87c2be8171a248a588a2885b79258ba6 : FAILURE in 32m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a0b0f92597d4015a81c87e23070e88b : SUCCESS in 1h 04m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4522e7e2bd4fb39e3f3e8998b42014 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7a2033cd3637eab8a52241d0a2acab383b97f4bf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-04 21:06:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"21fba708e5a2ec5e5ef9fc79344d8bdd71028667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 22:48:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7773fd0bc40f4ddead5bded8d7ba8a95 : SUCCESS in 8m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9115565aeb5147959e047185d809749a : SUCCESS in 1h 14m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/782c5da7875d4fa586a1fd6646cb4fd9 : SUCCESS in 1h 21m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/65a872bcdd054c13b9abfb38b1dfb7c7 : FAILURE in 24m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/581e304cb5414950a1198042930d03db : SUCCESS in 1h 28m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a5b666cd5d848458995833617721634 : SUCCESS in 1h 15m 35s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/af4c7b071f6b453684b3160777234f7e : FAILURE in 27m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f019a316d09045588c9a520acd82059e : FAILURE in 1h 06m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bc541c26f09f4bdeaec3880395f3aa9a : SUCCESS in 1h 15m 03s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3f43d81dfd6742cc9c30a408f358a78f : FAILURE in 36m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0c53203d9ee412ab239c9db4e879662 : SUCCESS in 1h 05m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7890b8949d7743b9b98f6b38a646b9d8 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"033c2ddcc1df4973160b1c181c1b0643942210a0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-06 18:10:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74ef17c814c432ae9d193ca2c0ca0c63febf0c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 21:15:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/27a4254266934ecd989476cfed23225d : SUCCESS in 9m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e14a0dd1bdd74c10b0177319f537ee9f : SUCCESS in 1h 14m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/4e5b6b0007c94d20bae99bfc6327d479 : SUCCESS in 1h 22m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fa6a52704999428aa554b7bab9313830 : SUCCESS in 1h 46m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2723d5ebbae54624800c82ad13b97f45 : SUCCESS in 1h 34m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17e6b5fd186c47d0b25d09fc41440001 : SUCCESS in 1h 18m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6ccc990d8b344690b7af4f315031a257 : SUCCESS in 2h 38m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae1630d8a142406da34294b86ac2ee0e : SUCCESS in 1h 56m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9923e9b339f74a99ae23a8c5ae919930 : SUCCESS in 1h 12m 58s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/113ce2026b4d4d86afacd73ee5cdf3fc : FAILURE in 35m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f2600fdad6c4390b5a412bded00d423 : SUCCESS in 1h 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520858221381495a9a2bb225bc7aa313 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":1},{"id":"fb4d6c0d6494e1670198ba2709246549ff490680","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-07 07:32:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6753dbe83f59af238d68d8ead6c68511547d8171","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-05-07 11:25:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cd1450f63a50d21c73cd1a7149391ffc26e6d31a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 11:25:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7630c34f9f6592a5c1f14b9b1ad603e3165719b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:34:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7ffb58613152430db50791e8d1d44426 : SUCCESS in 11m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0b980652f4304e87868d95d529509570 : SUCCESS in 1h 22m 44s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a9ac57e18ac34bc08bd56faf743454e3 : SUCCESS in 1h 27m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0a7b242edd124135961b140c15cd605c : SUCCESS in 1h 59m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/43392f2287c34f5f9aad7ed15c175514 : RETRY_LIMIT in 14m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/329e7adc12a745fca00a931fc6a2c652 : SUCCESS in 1h 21m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/117a9568e93c4f8ebe37f66443268ca0 : SUCCESS in 1h 55m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2793d4ef2cdb4f6fbec55586a01f1af1 : SUCCESS in 1h 30m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/648c231ebbad42e48a9d21cacd682dca : SUCCESS in 1h 05m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e52e34a05646bea8dc0180215ba2b0 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"cdf916909ff4842116eb29b4bbacb87838336354","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-07 15:27:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"79f2bc5c6d69447fa8f9216f65784cff5b6c0c92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 18:02:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/97e007b33c0e4d2e861179b5fcf73186 : SUCCESS in 8m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/93e8610d89f14fa2ac124f06c15805fa : SUCCESS in 1h 13m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/eef75f30ff974b4aa14ae9e13f7762df : SUCCESS in 1h 47m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/93fd0c7b34bb45f58d1c3ffbc6686201 : SUCCESS in 1h 42m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/15ddf3797243464c9fdfe93c9991ae9b : RETRY_LIMIT in 12m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/011137c03b134d9dbfc68004808207bd : SUCCESS in 1h 38m 46s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5b089075749742c9ac0c0b7dff1a92bb : SUCCESS in 2h 30m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ab484a15d254681ac2a41d50d90d17a : SUCCESS in 2h 13m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f1cb0f12ff2f4549abb35b47ebcce622 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6c3f89b029fe4c5090b558bad7a4b396 : RETRY_LIMIT in 12m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/864f684449704d8491c870fc4a70e6ce : SUCCESS in 1h 08m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f9af03a19fe45a092a830cd5d01612a : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"38a4b7bf258faf9d934f0c0a5c7ccb86e561236e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-11 10:47:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a8f0dd8edc14c7d88462efacefaf585c58c001c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:46:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4b50d088c3fd4e8f84a653478e5d81d2 : SUCCESS in 6m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8325db17541a496083752c876b4a9c1f : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5a0d9b7feec1492cadb28facd0ebe75e : SUCCESS in 1h 20m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/08c3f769cd0b45c6958ddb7ae99e3d0f : SUCCESS in 1h 42m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/928c26a0b6ca4a48af4b6d0681e7d0e0 : SUCCESS in 1h 38m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aaebaf3d8d26486f849bc97d3683b344 : SUCCESS in 1h 16m 48s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/94af68b345b141ebb479fcb6dedbf78c : SUCCESS in 2h 45m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6027fb9b994643db9b92a76120917da3 : SUCCESS in 2h 09m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b1c5ed2cd5be48f5be6113c491363cd5 : SUCCESS in 1h 27m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7cee63cb29ab432a97db73e3b3da5246 : FAILURE in 35m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec53399065484094828e78b4d0957121 : SUCCESS in 1h 06m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b08cf23157495c9a38df698e2aaefa : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7df67161f752cffa42be1651b405b25e507a16e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:46:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66778283a924075e29516490421f24bdf0e7fd49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:07:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d21cf32fecf24d30a81684d8419eb88d : TIMED_OUT in 18m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/17eb076a843346d2a335b7c06b8e0e01 : SUCCESS in 1h 12m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ac1ede692db8418c8a046a66cdf2d483 : SUCCESS in 1h 23m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e52269db124d4b9ab10858309fcaac6f : SUCCESS in 1h 37m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8b143ba92a264ec2a0533b656c27d838 : SUCCESS in 1h 30m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d120a81c46e48e2b8b47497ce5420b4 : SUCCESS in 1h 31m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9760917ec9b2421989aa56bf8a497991 : SUCCESS in 2h 01m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/900b278376be43ea9ecad3c50c74a6b0 : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c78420f84bee4063886ca2be43e1a31c : SUCCESS in 1h 20m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6345caa339bd4ecf8c52599e2dd867be : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a7a96fd7d2a06f2a043cb5a2ff1e84c4e63e5e71","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-11 17:20:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b11b16bb90234cd2e9ce252cfb64b320ab9a7c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 20:18:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/32a236f380c246a382da011d9dea8cf2 : SUCCESS in 6m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/88f48127667246f6a9677bf88aa3f86e : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/0fb37ce5f4bd4a0b9f74f0f4fd0f4798 : SUCCESS in 1h 32m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b6459bf3774949b5871ec7919ec12024 : SUCCESS in 1h 44m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/88133437bd5d4e18b119e458c76ced07 : SUCCESS in 1h 28m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/444ad46c8bdf4f48bdbd4270e5d22309 : SUCCESS in 1h 14m 10s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2db3a0c40bb5422ebeada02b3ef1f2a5 : SUCCESS in 2h 47m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6bd19de1b424e6c8985e84b6f09d391 : SUCCESS in 2h 03m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1b458263e43f4345b72760c135a407e6 : SUCCESS in 1h 29m 56s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6981334854ee4db794c95f033e2fbc40 : FAILURE in 33m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fa6bf01ca174f3786c61905dbc7c936 : SUCCESS in 1h 09m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea1f28cd3ab4a65930cac412c2b8fb0 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"49a6eec689343258d4efaa86f1e1ee1e1b1e7183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 20:18:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d94e6cc685148953e3201826830cf31313a5d7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 22:37:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"418e1ff102483b5f02f5d276aa4630de6699674e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 22:37:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/35ce5f7dd58241628e0971bac8345161 : SUCCESS in 9m 17s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d4e904113e1645018a9cc03094a773fa : SUCCESS in 1h 19m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/85db0d92bdef4f96a4152294e97ff1a3 : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cab19a858cd54a36a517928987432676 : SUCCESS in 1h 37m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b11cd9c1aa1d4c918f2acdaeeb908aa1 : SUCCESS in 1h 32m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae97344b5f1e42bcb77970f77f6b28d6 : SUCCESS in 1h 22m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ae7fb91619546b4a5534cbd8c65258c : SUCCESS in 2h 09m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5071d7e8377141b789d641d45d47255b : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0afe93a41c034e72a0b6526d15e97915 : SUCCESS in 1h 04m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20dfd233b8a44bc6ae20af568f47c803 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"319edb0425b5dae047a19aa58f65987d568c8155","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 22:39:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/320c75ff046147c6a4c41bdcf19c7ccf : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67cbc530bbb9bf4fd95542683b1179bf068bcb5d","revisions":{"67cbc530bbb9bf4fd95542683b1179bf068bcb5d":{"kind":"REWORK","_number":1,"created":"2020-04-26 21:52:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/73/723273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/73/723273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/73/723273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/73/723273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/73/723273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/73/723273/1"}}},"commit":{"parents":[{"commit":"bd265c880f1f6a2c0200e496cfe24e556fae7bc1","subject":"Pass _service_region to service_setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/bd265c880f1f6a2c0200e496cfe24e556fae7bc1"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2020-04-07 20:07:20.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-26 21:52:21.000000000","tz":0},"subject":"Revert deleted config section in Swift config files","message":"Revert deleted config section in Swift config files\n\nA recent commit [1] removed config sections related to logging that\ncauses various Swift services to fail to start. This patch reverts the\nrespective config sections. These may be removed at a later date.\n\n[1] https://review.opendev.org/#/c/671711/\n\nChange-Id: Ie73811e36f1ea35f060cb5af4e1275e7e69d4179\n(cherry picked from commit 91f26a98cf067a963ae7dc4809ce1438b70e0279)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/67cbc530bbb9bf4fd95542683b1179bf068bcb5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/67cbc530bbb9bf4fd95542683b1179bf068bcb5d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
