)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~723274","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Ftrain~I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9","project":"openstack/openstack-ansible-os_swift","branch":"stable/train","topic":"osa/swift_replicator-stable/train","hashtags":[],"change_id":"I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9","subject":"Fix replicator services","status":"MERGED","created":"2020-04-26 21:53:00.000000000","updated":"2020-05-10 22:49:53.000000000","submitted":"2020-05-10 22:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723274-1589150906662-11ac1cc1","meta_rev_id":"ff3a51c9fc98b7394b89cb37b0b24efcb8aaf251","_number":723274,"virtual_id_number":723274,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-10 22:48:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"recommended":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-09 23:19:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-05-10 20:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-06 21:15:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-05-10 20:51:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 21:15:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-09 23:19:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-10 20:51:19.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-05-10 22:48:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1731d96fcb6dbb7411b0e5370201c30fc953388","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-26 21:53:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"35d89ba72b87b0beb6e525960ec945c5198874fd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-26 21:53:15.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit e54ba6fe3166e09a77abeb0fb4d9b88b42630cb5","accounts_in_message":[],"_revision_number":1},{"id":"0cc75bfbd6a59bccc1608e94e4f002a34899cc8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 23:58:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cc29fd1d954249a9b4aa9a4010a6358f : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e1628ad72419448c9e19bb37aca6130c : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/26070ff7a1cf40a8bd3aa1f6c648ccf6 : FAILURE in 1h 03m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2ceac03374b14780a191bfe6fbbc0d1b : FAILURE in 1h 15m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2f0c2703b6ab4f44ae2e971c187cce2f : SUCCESS in 2h 04m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cf2c01c4fca4b7fb6c523c31b1d194a : SUCCESS in 1h 23m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0526f8a7624d42b4950ce275f8e001d4 : FAILURE in 1h 58m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5f09dc981f44f6aa6877a9f04fe4ca6 : FAILURE in 1h 01m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8303b9f955d043d5b648e670a666d7cc : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a1f460b2362a4188ae6cc5f48b162f6f : FAILURE in 35m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f9a064998c343bdb7628028f128277e : SUCCESS in 1h 00m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3032f326fa407a9917a657c9bc1f16 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4eded4c64629a3ed7862d5ed54053cd354204506","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-29 06:19:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5b9fb3ae4bca9b0ec059c4ed740e1de6d95b0fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 09:34:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6f28a2a37c1547b09471e44d125f9642 : SUCCESS in 9m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d5b6ce6991ba4de99422a8269aaa2478 : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/808c7defeebe49eab6d8d008c2e80e04 : SUCCESS in 1h 29m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b887370cf26540e0b49a79e6790799e7 : SUCCESS in 1h 44m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/eda41a3dcb7c43c4a95e51aa63fb7c96 : SUCCESS in 1h 37m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31c4f49faa2c448aae8306684c202158 : SUCCESS in 1h 18m 26s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e81a5a7147cc497d8464c799160bcc2c : FAILURE in 2h 28m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6aa85c4032bd4a15b4e822ca1a5411f7 : FAILURE in 1h 11m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/30706e59b11148e4a63169c01167ad8c : SUCCESS in 1h 47m 06s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3f191a9f0f1f43eb8b1cfa49d025bfdb : FAILURE in 42m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/017b5664e06149b9b031670698de2168 : SUCCESS in 1h 06m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0edb33af8d4beeb6a43736195d90af : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c0e373c94bbc3e1483979cfe03735697fb5ce2e3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-04 21:08:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7534e4f4e6d9831a4472cc885986e33c9bcca9f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 23:16:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d4cb8d315d6546bfb9fb12c9d546c3a2 : SUCCESS in 12m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/31e5caf911c747b79fff94857072196c : SUCCESS in 1h 11m 54s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3b977b22569340d29573a74d983172b6 : SUCCESS in 1h 38m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/18a26bfb490d4c7fad79d2a49ba697ac : FAILURE in 31m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/55bbae2ad1f347c88e3c04dac46923c6 : SUCCESS in 1h 43m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d29907d422244bba985e4fa3769bcf6b : SUCCESS in 1h 21m 28s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f8f814ff73d747e2a1e3cf3ab5756c27 : FAILURE in 29m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83d4511a04104a478a5bb4654b19fb79 : FAILURE in 1h 07m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dadeea3f96b447d986e309ba1e72d524 : SUCCESS in 1h 25m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f2881785897641c581f0c9ee05b24115 : FAILURE in 43m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19b2ec8f27244fb2bba935c4e7ab2a33 : SUCCESS in 1h 08m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a047afcd32534dd2956f5d89af6f219a : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e965335118b2f3b03993669bef4015a5be4d9f26","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-06 18:10:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bddfe26079e2f95a29cc2eb22cb06aae3ee259f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 20:49:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6b4984df57344c13bb47ce03ed36d154 : SUCCESS in 9m 26s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dd78f53642e14ca5a4776c3e99049f35 : SUCCESS in 1h 17m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/1fbd1ef6e305408b8983e695b341d076 : SUCCESS in 1h 37m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/455ba8071c2146fd98ebd0b2e1150dbf : SUCCESS in 1h 45m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/61af2ee2a94d411d9d6c66671c248495 : SUCCESS in 1h 31m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/839fdc5dff544bc7b950e12c4637feda : SUCCESS in 1h 20m 51s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ea8d8831338e4d9f845a6d8fb1ad423a : SUCCESS in 2h 32m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff10f68741bb4c728f8723ef3f6375f3 : SUCCESS in 1h 52m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cfd27599510b46eab858087cf56c952d : SUCCESS in 1h 32m 52s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/59bb3aa00f9140abbc22781587d8f98e : FAILURE in 44m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c310b8294eb74d07b9a1719c4dc4d97c : SUCCESS in 1h 06m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/712255e3e6a546cf9b07604eb1a06eff : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c2699f05ea6abbed612ec784b2443080036d8cf2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-06 21:15:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"efae689f16be7a78d845211ce8ea513c3df634e9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-09 23:19:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4466334bd2032c6db7da0544db70e9306af8b50d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-05-10 20:51:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b54c7db678e8c01ba8739caaad3f89f7b8385349","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 20:51:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5deedacbebcbd80ba4c14622e3bc3bc1f56945af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 22:48:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3acc9178dcb5543a084f47877c27f793d70e02ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 22:48:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0c0e8d6cad914698a7fec5c00f0891cd : SUCCESS in 6m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/eec79843d0e14d3fa86f352e9859d0e0 : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/aea2ebe359d745d4a8caee43aad6aca9 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/324b56340762437d9359e32180d1f387 : SUCCESS in 1h 34m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c09e634406c94c54b64c0ab4f07a949d : SUCCESS in 1h 32m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55347a1894ea45a78f478ddd22082a1d : SUCCESS in 1h 17m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/765d6882dd264455918649387ffe40cd : SUCCESS in 1h 50m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/80a5048984ae48b28ba2715d9a779ad0 : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccdb9033ab414a76a168d8cb538c219a : SUCCESS in 59m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dadb81181fab4b94811b41c3852402c1 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"132aba9c94fcd1c2b268cfb3e49076a569598979","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 22:49:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04506d59a2cb41bbb6b568f58cc4c5da : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a83c9183edb69e1809009720b2e1f2eed3a4a93","revisions":{"8a83c9183edb69e1809009720b2e1f2eed3a4a93":{"kind":"REWORK","_number":1,"created":"2020-04-26 21:53:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/723274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/74/723274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/723274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/723274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/723274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/74/723274/1"}}},"commit":{"parents":[{"commit":"bd265c880f1f6a2c0200e496cfe24e556fae7bc1","subject":"Pass _service_region to service_setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/bd265c880f1f6a2c0200e496cfe24e556fae7bc1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-08 17:13:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-26 21:53:00.000000000","tz":0},"subject":"Fix replicator services","message":"Fix replicator services\n\nCurrent services were infinitely spawning replicator process\ninstead of running a server with another config file\n\nChange-Id: I3310c11c0be38ae72b1f0bf94a849587fb8e5cf9\nCloses-Bug: #1859159\n(cherry picked from commit be9ad06226d74cfcb796834d27ca641902329fcf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/8a83c9183edb69e1809009720b2e1f2eed3a4a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/8a83c9183edb69e1809009720b2e1f2eed3a4a93"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
