)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~745863","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I5ae8285aaaca73ca9dacff18dcf31e4754b5742c","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I5ae8285aaaca73ca9dacff18dcf31e4754b5742c","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-08-12 11:35:25.000000000","updated":"2020-08-14 21:18:44.000000000","submitted":"2020-08-14 21:17:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745863-1597439822753-69ff8aaf","meta_rev_id":"92f84fe5ed46c4e6dab5840e76d82007fa3f0e0f","_number":745863,"virtual_id_number":745863,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-14 16:51:52.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 21:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-14 09:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-14 09:31:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 09:31:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-14 16:51:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-14 21:17:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa083870939e5b71e5d498c8b38111ac3c6624ff","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-12 11:35:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2351080e97ac2762fab03b8249b1bc3ed7a7148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 14:32:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aead231c2e4b4723bc34b56da4631e34 : FAILURE in 10m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d9b5dfc2c6e14db8b39528d8e5a32749 : SUCCESS in 1h 10m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2de715a6d37f47bbaa9b5c6b85d90b34 : SUCCESS in 1h 26m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d4b89d4c9ea8415fb66588bfb6571cda : FAILURE in 1h 11m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c277a6e6e60c40f2906f07c7f3c41b22 : SUCCESS in 1h 06m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98ff83daf450422d96e10802f54c2734 : SUCCESS in 1h 06m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cb169c6e823b45f6b33a85b6db99a899 : SUCCESS in 2h 25m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/882fa777456943039876da58f75d642c : SUCCESS in 1h 51m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/41cbe1a8e1834054afc8169a42de253d : SUCCESS in 1h 01m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/36c956b6123d481eb82b68ac39755263 : FAILURE in 45m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/180be7c6148348a59d6b871aa6891f0d : SUCCESS in 55m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/439ba40549674ce8bd47580ec1c6d86c : SUCCESS in 53m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19fcc60848514fd1a98588f76b54723a : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"29c2c7befcf5141dd1c4b49a706a221d40c911da","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-13 05:44:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11825441ad19ab3f89df587dcfe02cc38d3bb339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 08:54:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f4185a616365409e8bca8a53822ba0b2 : FAILURE in 13m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2fdf12d466c6417da30d3cdeb9a44732 : FAILURE in 21m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ab4a31a85d2747b6a0891e8b002da265 : FAILURE in 1h 35m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/716d1b0474774c71a3ba307b173f95da : FAILURE in 1h 30m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f323fc89f9824f5cbac01d15073c723d : SUCCESS in 1h 36m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f76b63c14a3a45428ed6b16e33ec1d9e : FAILURE in 1h 28m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6de96b51b687485bbd4cbbac3d3716c7 : FAILURE in 22m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a7c012cbec14182b4b6e1bc4999b5cd : FAILURE in 20m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ffa903c868a04eb3b994974047d3cfd9 : SUCCESS in 1h 07m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fbf081fd9ce04db191ba44e31196f296 : SUCCESS in 1h 00m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56737ea2e6fc4df4a581d7c2fd9c26ab : SUCCESS in 1h 00m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a64e766329dd45c19f0b01df78045cb8 : SUCCESS in 55m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ed84ebaafd4b10ba47f5ffa5821f58 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"eca81a64e1c61f8b9086647ce0466f13b0756b98","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-14 06:17:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7178d1554733bdce9b5568916923b4dee5237505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 09:07:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8ece7dca15a34b85ad8cdc0349b76902 : SUCCESS in 7m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/721c3979ecac45619080c0afea47d8a7 : SUCCESS in 1h 40m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d0dd1c4d226343dabf31e7c54ff6436d : SUCCESS in 1h 43m 39s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/090ae97f986a494499b15e5818b6b967 : SUCCESS in 1h 19m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f83dde191dc43d7b545dbbe6fd1d35b : SUCCESS in 1h 18m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5730417d82b847389418d2471e37ed44 : SUCCESS in 1h 04m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b00498d8393e45ccac5e7acae246ffe9 : SUCCESS in 2h 26m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64382bde5518401f8ea20663c4ee18f7 : SUCCESS in 2h 03m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/996397eff2a243a2a085387c2c7e2b27 : SUCCESS in 1h 01m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/df313ddc81884166835381b4258a301b : SUCCESS in 1h 00m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcee4d7df87b4be8818bfaa32f401b5a : SUCCESS in 59m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2258eab1c3484cafad128174f3046106 : SUCCESS in 54m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e4b90cbbac4bd786361ab84a16ca18 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fe6847b08e5a17b3549b6b37a83e0b7b1d2044f1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-14 09:31:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f66cc56da819bf8169137bd2da2a527998755eca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 09:31:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"173892c32842a058de1f14701e65c69b48251cbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 11:24:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e80107f084fe460d997c7abbdfb06eda : SUCCESS in 6m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b6fd26bddb204a04909252e726490a4d : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/62566e335c1f4dfc99a292f840e881f2 : SUCCESS in 1h 39m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/01af6367569247d29888e65e58f0b6e0 : SUCCESS in 1h 21m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64c93dc48972469490a7bec62f791b42 : SUCCESS in 1h 01m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1d73ec07141490a825f0fc8a4678c9c : SUCCESS in 51m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6893714ac8d446a8981a102885b1dde3 : SUCCESS in 1h 02m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ec74627b25f14009aaa80fde507a3d72 : SUCCESS in 1h 02m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dbd2605333847b9a7767de42f9b1402 : FAILURE in 23m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cddeb76b32214a85b1d810c50fe05e1a : SUCCESS in 52m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f328153c9ffc40f2ae44790bce844785 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"23900f1a7ca66d40a6d44dba9043cd1829565de0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-14 11:42:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bf98274e8fd297c540c39926308942a625277ee9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-14 11:44:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7da556432a43ddbe6ae99e149fef2175c3792ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 14:52:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/40ce33a8812e44a398d735bdb6175908 : SUCCESS in 6m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/90108794e049430f841792823fdc36d3 : SUCCESS in 1h 08m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/09c71bcd00a748c58566866af8265d96 : SUCCESS in 1h 30m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1ff6d1dad3ba4e0da25181afa4e8ba38 : SUCCESS in 1h 23m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73f6a5191b6b4be798807400ee6827cd : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06574ac7e9bf4ab29f416a08f21741ae : SUCCESS in 1h 06m 11s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e9ed7192110642a0981476c0f4aa811f : SUCCESS in 2h 53m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/128916d1f97e4fed8b28b58963deabf8 : SUCCESS in 1h 53m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5235703e2ba946d482b30089e372b0d5 : SUCCESS in 1h 02m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/093d13bf8f3543e0bff91ef7c2740410 : SUCCESS in 55m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/531f4e616dfa439abbc502199a001a7a : SUCCESS in 56m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00af64c112bb42ada2540e0694782c20 : FAILURE in 1h 02m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6562d516fba4dabbc02dbdf0bfcadd8 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c31893facdd0794c45bb809d7b560ea2954b391c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-14 16:51:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8364c381b6b92f4522c5e1c5a6b441407f6fa596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 19:11:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fe3fb5f178144f198254bdb7f5506382 : SUCCESS in 8m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d89f41ca46ba48b19d0444d81df23958 : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d9b292768f934d72998c338d434682d9 : SUCCESS in 1h 17m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e86596264c8476a96cdfcf82cfeab57 : SUCCESS in 1h 46m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f119e17b9104d06ac086825ecbcdf17 : SUCCESS in 1h 08m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6c44c07bc8d470f8ef035692a3474cf : SUCCESS in 1h 13m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2d106c07b9534a54893b03ec7f6f5617 : SUCCESS in 2h 09m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43e140e24d894c23b356b739e4d5e4fc : SUCCESS in 2h 15m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9c1a6d32356b4b1ba50c9fba12970c98 : SUCCESS in 1h 02m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2757dc1b5ec8490c8ceca033a9f30e51 : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71e9b0d0119241c9ba6a7fb50542987c : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/686e06b7f8064b86ac9d43d4cf6b8ccd : SUCCESS in 53m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46e50b46f374414e9c2bc83ed8f75208 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"639ecba9f6cdd5591f9ce255175b01d58c7e4f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 19:11:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee238f84c47c573d9348cfe638d24fc92ab2bfd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 21:17:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3eb20011e792406cbe0487282d7bc0bf : SUCCESS in 6m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/acf1d076fe7b42a0b2a6d92aa0da5dca : SUCCESS in 1h 01m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/81cf680147914fb5a629b303f896f9f1 : SUCCESS in 1h 45m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a2b337508b9462bb48ef03f16b2682a : SUCCESS in 1h 59m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2722c5f2e6904d039fbed82416acd4b2 : SUCCESS in 1h 57m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3d4df870f624005a6dd3bcfbcc842af : SUCCESS in 2h 01m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8ad1cdff4a5842768d4bd368da84db09 : SUCCESS in 1h 49m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/31c0d917f38f49c7b0f090372c577e33 : SUCCESS in 1h 45m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7673abc15974b82ad9608189f2abc56 : SUCCESS in 1h 42m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/124f3373fe034c859992247d99584bbb : SUCCESS in 1h 35m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd8f50ae8fb643989cba2cc2f2b07d7e : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"39c0f975db4b271b53c5d516cbc3cda02b62bcaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 21:17:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a5244e283befce7454c681c14a143c7d6ead239","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 21:18:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e09d2d92a14431a26314b3a0386e42 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9924d36dfae6a3c7439823bc20491788801d1f74","revisions":{"9924d36dfae6a3c7439823bc20491788801d1f74":{"kind":"REWORK","_number":1,"created":"2020-08-12 11:35:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/745863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/63/745863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/63/745863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/63/745863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/63/745863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/63/745863/1"}}},"commit":{"parents":[{"commit":"abb1c721e50258a0fede787b810e3b17e59d85f7","subject":"Cleanup ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/abb1c721e50258a0fede787b810e3b17e59d85f7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-12 11:35:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-12 11:35:23.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I5ae8285aaaca73ca9dacff18dcf31e4754b5742c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/9924d36dfae6a3c7439823bc20491788801d1f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/9924d36dfae6a3c7439823bc20491788801d1f74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
