)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~748217","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Ib8a8875c1a355b1f67cdc2bc4f1d028fa7ae3496","project":"openstack/openstack-ansible-os_swift","branch":"master","hashtags":[],"change_id":"Ib8a8875c1a355b1f67cdc2bc4f1d028fa7ae3496","subject":"Delegate gnocchi retrievement task to setup host","status":"MERGED","created":"2020-08-26 12:50:37.000000000","updated":"2020-09-07 13:07:00.000000000","submitted":"2020-09-07 11:13:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748217-1599477181664-33708db1","meta_rev_id":"f23a9a2ab548fae94d71f0f7e6881a20bdd4acbd","_number":748217,"virtual_id_number":748217,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"date":"2020-09-07 07:04:21.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-07 11:13:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2020-09-01 07:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-26 15:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2020-09-01 07:46:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 15:28:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-09-01 07:46:15.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2020-09-07 11:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7202bcb7256c7855ecbe1ca8827a5709dca9dbfd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-26 12:50:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac6741dacb94cfe421b8561675045f480353df9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 15:25:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/202a8790df604d049b518d285b3b332e : SUCCESS in 9m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dd4b968f1f9f493e841882fa5f97ccae : SUCCESS in 1h 14m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5a1cfff1f9254f51a87157699d58f3c8 : SUCCESS in 1h 36m 54s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/34092bcc052f456cb2bb491ad8aaa73b : SUCCESS in 1h 32m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/123fc7c1670a48fa8a8e0e6553b69241 : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e54a523c2d9420ab8ed393d88690c3e : SUCCESS in 1h 02m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/169656eb5a2540339bfb9e0486d4aeac : SUCCESS in 2h 16m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74221c485f4041e4acd3643f4bf99824 : SUCCESS in 1h 54m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a091789e60d54dab9658138b5ffd0712 : SUCCESS in 1h 14m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1aed7fe3f8cb4e719e1c2d97d1f7f382 : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2eeff7845af949b2aa61a7d5bb333f7f : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24c91d5e152345469e2a177dc698fedc : SUCCESS in 1h 00m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae7378b169b64df5adbcb5bd1ebd3bdd : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ea1d5750f30a24695330413d3179f0451e401034","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-26 15:28:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3e4b7ee40c936ee13f7fd150d4e98cedc8ff27b","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-09-01 07:46:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24fbf75c3a975894bcf37f9a3fd1859cc9f1826e","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-09-01 07:46:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"88dff8c0f084ed1585cbdec0bdcce06995c92a67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 07:46:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d7f428ed2d32c0475352815c72151269db6e37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:36:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8fcd84c8f3b843bea042a9d7f6eefb03 : SUCCESS in 7m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6fdd682f864f49a4a1a323c0d3705f31 : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c12dbe3dfda442c48fae3ba690011225 : FAILURE in 1h 37m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8228b2551c7b4bd1a1a49b8888d99c09 : SUCCESS in 1h 37m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4495bdbc0f544ffbfa106ea6f4e7468 : SUCCESS in 1h 17m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caee7a7f890d4e86b90155afce6634c4 : SUCCESS in 1h 10m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c7c7dd5b39d24926b60fe86d2f233974 : SUCCESS in 1h 07m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d37df2d51c8544018a9a11c459dece77 : SUCCESS in 1h 13m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e125daad971426c83be532ecfe3e002 : SUCCESS in 1h 04m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f425039153d4966817d42bbf4b49193 : SUCCESS in 59m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6f0fcfc86748fdb65dd621100255b3 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4b8a56a9d1f2b9477b7926016db224fff0717a6a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-01 10:58:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9b660084f5af1292764c8439b11ac93ed740510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 13:21:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/83d16cf454884f8aafcbb0b6385cd667 : SUCCESS in 6m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8a29ff2997dc4824b39d289df1e2b340 : FAILURE in 1h 08m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6728f1beb6d941048a8c0146062b9646 : FAILURE in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e0bfed70078c43baa0a635e556d96716 : SUCCESS in 1h 26m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88847a0ad09343da9a108e885373bb11 : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9db32b99d0f422cb980a4d85b6ae0ef : SUCCESS in 1h 11m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/380cad83e6d74c1fb58fbd7647b662bd : FAILURE in 2h 18m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36ca61acaaf349c29cba24a6b163bead : SUCCESS in 1h 52m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0f6114f513dd4e79ab3bdfbb90dc0843 : SUCCESS in 1h 07m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9d0756eeacd445828aa84120c33beb13 : SUCCESS in 59m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/865c0120c8fd4116a01c12298700f7a4 : SUCCESS in 1h 04m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d823126b88e14acaa420f2535ca4b638 : SUCCESS in 57m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7568ccc4d4497498d440846cff7839 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b45f70815ab1d0fb5868a23b8237ab5458ce63e9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-01 14:06:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26294921cae24d2d3d339d58db1df58cf69c9376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 16:30:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/59c67045cd95439c8a542a9222f425f9 : SUCCESS in 6m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3a729ef1fd244ba9822b61ad883268b8 : SUCCESS in 1h 21m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d8ab245542194b8cb18fccd5e4727bb3 : FAILURE in 1h 18m 39s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/963d5063d1784b9a825c0b084468384c : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf48ec067eef47f09b17406024a712f4 : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/612ee8d04773446fb0eaf5a0b88e7a2a : SUCCESS in 1h 12m 03s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a6d0b5267bf14c83baa545d099fa07f3 : FAILURE in 2h 19m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7f269149e0044dcb934017c70e9999e : SUCCESS in 2h 00m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ad610a13b59a427cac6431f66a1dfb33 : SUCCESS in 1h 09m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/91e1cd54503d4fafb840258615fdce2e : SUCCESS in 1h 14m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b44486a8ce6a4f27ba3ed57d34a39c74 : SUCCESS in 59m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4844dfb446a24aa0a93ac73aaba7e92a : SUCCESS in 1h 07m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22d2677cfca4f679822cf8628c5d43b : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ae6bba74ced5f78469fd542b67d85360286e76d0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-07 07:04:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"453397c87c10b9f93db53f34c0f3b30fe674def6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 09:47:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bb96889543cd4f64ad884b34359efab0 : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7d50168d853b4e1fa3e40ac5e0b5f2ef : SUCCESS in 1h 00m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/437395087d584cd0854ac1a5b4a113dd : SUCCESS in 1h 10m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/921ef127b3ea4a8ba3c4e84230cece27 : SUCCESS in 1h 06m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4071903a94ab4e41bfae1a9061a4ebed : SUCCESS in 57m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e2cf0f33f4fe4a459f2f49625939c337 : SUCCESS in 2h 21m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/022aa976d1b34658b6883dcc4831cd17 : SUCCESS in 1h 40m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/802b478f7b7b45989497ae2e4d29018a : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bd82de5b7514cf88a79ac3b0fef6198 : SUCCESS in 56m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31b1e8939c944de685c9c68467f55317 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c58eac073afbd57c20479a2ca9e6c8d9d5cb9e44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 09:48:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76266e332cc3ff52916066d7e042582004f097dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 11:13:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/49ef488d49754ebcadb8698f134ca610 : SUCCESS in 9m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bc3b404a94a540a29ed1d9c1ccb2e1b9 : SUCCESS in 1h 07m 10s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c84dd4d5cf4a4e4fa780e01b5312d546 : SUCCESS in 1h 20m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6eeb68a3549463fbd96a797e63cfb0f : SUCCESS in 1h 13m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d73a712a13e46a79e65a3dda3188212 : SUCCESS in 1h 09m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/53ebcd7f2ac04a96bc42164805eee6de : SUCCESS in 44m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47efda516ad74b5bbfdbdcbed13ccfb7 : SUCCESS in 58m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9144116eae4ce78e4027acf815bee2 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a4b9fd06f0d7683787c857bdc2d24d503c73e0e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 11:13:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4e5ea66450b7afbb1ad9df7bf891e9ddc4feb66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 11:14:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a13e3f103914d76b7c2b403b2244ebe : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f23a9a2ab548fae94d71f0f7e6881a20bdd4acbd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-07 13:07:00.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 8521791539b19841b488ef71edf6b7c8e67e9d02","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c1209e0ec8b20869bdf6eb4d0a4ebba02264fda","revisions":{"5c1209e0ec8b20869bdf6eb4d0a4ebba02264fda":{"kind":"REWORK","_number":1,"created":"2020-08-26 12:50:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/748217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/17/748217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/17/748217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/17/748217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/17/748217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/17/748217/1"}}},"commit":{"parents":[{"commit":"267bc5736159a628fca103e2cb890bc87e8937fb","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/267bc5736159a628fca103e2cb890bc87e8937fb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-26 12:50:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-26 12:50:25.000000000","tz":180},"subject":"Delegate gnocchi retrievement task to setup host","message":"Delegate gnocchi retrievement task to setup host\n\nSince we don\u0027t have required libraries on the swift hosts,\nwe should delegate openstack collection tasks to the setup host.\n\nChange-Id: Ib8a8875c1a355b1f67cdc2bc4f1d028fa7ae3496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/5c1209e0ec8b20869bdf6eb4d0a4ebba02264fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/5c1209e0ec8b20869bdf6eb4d0a4ebba02264fda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
