)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~786790","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Fvictoria~I5498030487677523ae71ae2f9a13074ab4552204","project":"openstack/openstack-ansible-os_swift","branch":"stable/victoria","topic":"fix-fallocate-percent-stable/victoria","hashtags":[],"change_id":"I5498030487677523ae71ae2f9a13074ab4552204","subject":"Revert \"split templates to work around configparser bug\"","status":"MERGED","created":"2021-04-19 10:40:18.000000000","updated":"2021-04-21 22:58:51.000000000","submitted":"2021-04-21 22:57:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786790-fix-fallocate-percent-stable/victoria","cherry_pick_of_change":782956,"cherry_pick_of_patch_set":3,"meta_rev_id":"214fc4c6d7ab62c677037dc9ba429c6e1c143422","_number":786790,"virtual_id_number":786790,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-21 22:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-04-21 12:17:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-19 19:07:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-04-21 12:17:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-19 12:31:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-19 19:07:06.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-04-21 12:17:44.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"d741e7dac8bdeb189aba0cede8782df7fb309e52","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 10:40:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* templates/account-server-2.conf.j2\n* templates/container-server-2.conf.j2\n* templates/object-server-2.conf.j2","accounts_in_message":[],"_revision_number":1},{"id":"aa89cb245a3195fea0e024f28d564467f992ee28","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 10:44:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c0b43c6d6888b3bb80db11b598e054e22e85a68a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 10:44:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"5325551945d05fcd177b50ed3dcf657c208caabe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 12:31:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/42dc57c6c5504448a348d430fbe7b616 : SUCCESS in 9m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/31be3c9b09de4cc694b744357a5f07b6 : SUCCESS in 1h 20m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cf565ba3347f46ad91012e6df8095f5c : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d7aa37faef34b02abbd5ffce6b04600 : SUCCESS in 1h 18m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6fedada9dd14e87a264389f54ad06f9 : SUCCESS in 1h 00m 32s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/42a5e955913f4e27950be2bc00156487 : SUCCESS in 1h 41m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c441366bb8cb4a7b9a4a90ecfd37f64e : SUCCESS in 1h 33m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b7be331e1d0f41ecabeff3910d827533 : SUCCESS in 55m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a5f0741b67a44188b0656ed6b2d3c8b : SUCCESS in 54m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd941352468842b38d7482c954728e0d : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2},{"id":"112144f7ccce4ee0f9171a3eb55d066b4c06f02d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-19 19:07:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f422eca222f6bcd3062b5bb162d22ef6b2191897","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-04-21 12:17:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a274d6beba4eaf72a5bbdf055a9bd448f47009f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 12:17:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7b92e2ee70e56cc848592f50de17887d1b8a6d5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 14:08:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/983dc3c22e6f464c85357f16d7744748 : SUCCESS in 10m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1a0aca73f3eb49bda589ba02ca720581 : SUCCESS in 1h 19m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d55116bf2fdb4a2cba6140cc5f43f04e : SUCCESS in 1h 23m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c82ab8f454344bebb9a2fd939c476a67 : SUCCESS in 1h 25m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f40db8565d24066933bfd92b8c3831d : SUCCESS in 1h 16m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5067652700c543a0a5cf5efd1e0e5f29 : SUCCESS in 1h 05m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4892c8306c9046d49aa999395e5b857c : FAILURE in 1h 36m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fba3adcd7726486aaade7d0558ae01ca : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"559a8b17506853690838289191686e4d4d9f3736","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-21 18:17:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6bc4cf947453cf636d446861a01c93d580716aae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 20:24:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4f1ba64a56d24fc5aff379b3ae905860 : SUCCESS in 9m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/28ca83e21ca8411bb1ad0a8a7ff80a0a : SUCCESS in 56m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/976d3e7f64e74f35a00095632f231e22 : SUCCESS in 1h 21m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d211fdd9df484f8da1250804a1196109 : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61b9508da3524dbbb77fcf0aaab13b62 : SUCCESS in 1h 16m 01s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/aef9b9be248b46aea6a97466fd6027b7 : SUCCESS in 1h 36m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db94de78a38347bf92a328d3fdaa1e85 : SUCCESS in 1h 59m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ec0457bd0c3e49e0955df96f8d2427c5 : SUCCESS in 57m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3be73e8bc7154eec8773a202f277e238 : SUCCESS in 1h 18m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee611d789794c138fc6028a8aa896dd : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c8dc84eb7386eeb907dc6e9f6c8b969c36201403","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 20:24:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1a66031ffae7b4d847a0fb3f2b3dde88946b8309","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 21:18:23.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"318887a8208f2c4cca9792bdf228c9e496ae32f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 22:57:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/13f4d523fb1e4aa78edbbf11c64bd5e1 : SUCCESS in 11m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/254e084eee05436d9c32fbac24f429dc : SUCCESS in 1h 17m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d2682983c5a0413e96502cb777f36fd2 : SUCCESS in 1h 17m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cec321006209474d8d13418d33b87f92 : SUCCESS in 1h 30m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51d823c6744a47d58a69b67eca0742e7 : SUCCESS in 1h 11m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a8f8b85242c246ed919e45ee70559424 : SUCCESS in 1h 06m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ac2ee4caa3945f98e48cac030a5f9a5 : SUCCESS in 56m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/078015af42f74ad5b1aaf67035b1b6c8 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":2},{"id":"101e1326c658a2c5327bcc58bf83c33e302dae25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 22:57:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"214fc4c6d7ab62c677037dc9ba429c6e1c143422","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 22:58:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07007a135a414b139c814811d85e7e08 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c1e3547190fb2c9294b47499646ee04e20b600e","revisions":{"daf88b40e8a7ab1e00ae70e2e1deef20796c6f7a":{"kind":"REWORK","_number":1,"created":"2021-04-19 10:40:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/786790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/90/786790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/90/786790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/90/786790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/90/786790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/90/786790/1"}}},"commit":{"parents":[{"commit":"0ba3b6d970e3d695a040570d05d0301fb565e5ce","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/0ba3b6d970e3d695a040570d05d0301fb565e5ce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-25 10:08:57.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-19 10:40:18.000000000","tz":0},"subject":"Revert \"split templates to work around configparser bug\"","message":"Revert \"split templates to work around configparser bug\"\n\nThis reverts commit 3e151d97adbff78be1b146938374a0dced5970e9.\n\nReason for revert: Upstream bug has been fixed\n\nChange-Id: I5498030487677523ae71ae2f9a13074ab4552204\nRelated-Bug: #1872553\nCloses-Bug: #1921354\n(cherry picked from commit 24bbb72ac97bbd2683d0eae04b8ab4c3674005bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/daf88b40e8a7ab1e00ae70e2e1deef20796c6f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/daf88b40e8a7ab1e00ae70e2e1deef20796c6f7a"}]},"branch":"refs/heads/stable/victoria"},"7c1e3547190fb2c9294b47499646ee04e20b600e":{"kind":"REWORK","_number":2,"created":"2021-04-19 10:44:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/786790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/90/786790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/90/786790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/90/786790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/90/786790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/90/786790/2"}}},"commit":{"parents":[{"commit":"0ba3b6d970e3d695a040570d05d0301fb565e5ce","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/0ba3b6d970e3d695a040570d05d0301fb565e5ce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-25 10:08:57.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-19 10:44:09.000000000","tz":0},"subject":"Revert \"split templates to work around configparser bug\"","message":"Revert \"split templates to work around configparser bug\"\n\nThis reverts commit 3e151d97adbff78be1b146938374a0dced5970e9.\n\nReason for revert: Upstream bug has been fixed\n\nChange-Id: I5498030487677523ae71ae2f9a13074ab4552204\nRelated-Bug: #1872553\nCloses-Bug: #1921354\n(cherry picked from commit 24bbb72ac97bbd2683d0eae04b8ab4c3674005bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/7c1e3547190fb2c9294b47499646ee04e20b600e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/7c1e3547190fb2c9294b47499646ee04e20b600e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
