)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~791250","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I40d3178d1b3bd1234ac34c6fa1d0d2bcb7160720","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"bug/1879192","hashtags":[],"change_id":"I40d3178d1b3bd1234ac34c6fa1d0d2bcb7160720","subject":"Do not collect gnocchi project ID","status":"MERGED","created":"2021-05-13 14:34:20.000000000","updated":"2021-05-17 10:38:19.000000000","submitted":"2021-05-17 10:36:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791250-bug/1879192","meta_rev_id":"d9575fa5a92ca3c5ef1c92004aafa98b7e9c7e2f","_number":791250,"virtual_id_number":791250,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-17 10:36:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-05-17 09:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-15 10:37:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-05-17 09:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 16:54:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-15 10:37:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-05-17 09:06:05.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"72da4f27e2a854f8614e950f2869fed17f5e7f85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-13 14:34:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7740d7fd5d35f4e8d4badf19c45cc8cf97b0331f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-13 14:35:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"673efc933225f57dc3f65d851055a061d59c5bc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-13 14:39:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"8ac9ce6a5179a386dee88ccc768af25751eb5b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 16:54:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7d07e6d5ceac4f469106324ee33e7f42 : SUCCESS in 19m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b080681c8d094886a5d62f32b043039a : FAILURE in 43m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3ec3e7c992b44800b9162ca65aa37909 : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc08044fe07046298dc4e2a43aa859c8 : SUCCESS in 1h 30m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/375d46751db34f94a0ab95fb28b40288 : SUCCESS in 1h 28m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb28e03df27b4e1f8265227af8283ead : SUCCESS in 1h 18m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0f418e589a5045e0baa9c4ebb5f29c4c : SUCCESS in 2h 07m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/237914ca7aae44bdbb7f4d697dfcad77 : SUCCESS in 1h 34m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6429997067b947099eb783cbf128c171 : SUCCESS in 1h 03m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c64a1c0fe7a04bceba5bfd2702d82aa7 : SUCCESS in 1h 01m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5660abc172d74caeb01d325446dbb364 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c53af2d04e0743539a7e61eaa297ed94 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"b95c7159669b8f5bc2f14bad0ca976a4737ed2bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-14 11:43:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"274412f71a73404cd56385821cdf9ddd2faf2e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 13:58:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5d83246968e445318d3661f0e44ca93f : SUCCESS in 10m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1cb4aaa512564eb1aee990d45d9de31c : SUCCESS in 1h 27m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c598677891054e07a6c3cad47193ca44 : SUCCESS in 1h 29m 40s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/54d93301b89a4ec2b8726b456e45367e : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed85ed09270c4d789558589bee107629 : SUCCESS in 1h 25m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e34eac5958f495fa1ca4fefff1e8166 : SUCCESS in 1h 30m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a187c8d8ca9546208aa111ed35db11a8 : SUCCESS in 2h 07m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9f1eb0d1cb0438dab613cd86294c1d9 : SUCCESS in 1h 30m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fe3a9e1fb894de290ed39121f4d835f : SUCCESS in 1h 20m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14ecec94ff714bb585a4c9c27472526e : SUCCESS in 1h 04m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a51a0809c042f6a8b0f9b7efe6ea77 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c81903546f644b33b34efe73d3655540 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"9cc6a62ae8b49c54098fd1811f746b865a0842e0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-15 10:37:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ff2efe4683e9c292bf9c8ec252e0e5a18a0a29e0","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-05-17 09:06:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"413f06f51f491e54a08a5b98a0430e006be1d77f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:07:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b6dac3f81504a00046d1bd58988fd2774916714f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:36:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/be960574275d4a35bbb4d644a20ec17f : SUCCESS in 13m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/05a2147ca7194b41adee221a481efeeb : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9f0981ebfca94c23a4cb500efc807c50 : SUCCESS in 1h 25m 32s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/45f35e54e4ce44f489e6f210e9711769 : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/450af85e34f84e77919f5f1e9327c963 : SUCCESS in 1h 13m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af05570434614357aaa8505af4f970ae : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/661954d5ec9646f19218018ed3935aec : SUCCESS in 1h 02m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34a6da7b41364c0e9a245100529d640c : SUCCESS in 1h 00m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1c79d6ade8e4e12966664c39d3b9e77 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/728d479aaa044416961192fda19b0f60 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"25245ceb80042c660ad455b96daf15b45b2cf2e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:36:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d9575fa5a92ca3c5ef1c92004aafa98b7e9c7e2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:38:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414ea3149d9144b8835b9759669996b6 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2ce00d6c69854a175d876c95b3439863a20a5fe4","revisions":{"3a98ef30cbb8770ffe1cce3e29b690a75af81475":{"kind":"REWORK","_number":1,"created":"2021-05-13 14:34:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/791250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/791250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/1"}}},"commit":{"parents":[{"commit":"d425a9486551c717e1e471e7f1ca62babe6f9b2a","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d425a9486551c717e1e471e7f1ca62babe6f9b2a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-13 14:34:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-13 14:34:08.000000000","tz":180},"subject":"Do not collect gnocchi project ID","message":"Do not collect gnocchi project ID\n\nCeilometer middleware now accepts also project names,\nso it\u0027s not required to gain project UUID [1].\n\nWe also switching gnocchi project name to `service` for all scenarios\nas it\u0027s already whitelisted by default [2]\n\n[1] https://opendev.org/openstack/ceilometermiddleware/commit/e2bf485044d8f3743da9298a9e461c5808be31f3\n[2] https://opendev.org/openstack/ceilometermiddleware/src/commit/082dabb1b7186f614664f69db32f80918b20d624/ceilometermiddleware/swift.py#L158\n\nChange-Id: I40d3178d1b3bd1234ac34c6fa1d0d2bcb7160720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3a98ef30cbb8770ffe1cce3e29b690a75af81475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3a98ef30cbb8770ffe1cce3e29b690a75af81475"}]},"branch":"refs/heads/master"},"37fc6b9160856c27d9189c769297e2f9358bccf7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-13 14:35:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/791250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/791250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/2"}}},"commit":{"parents":[{"commit":"d425a9486551c717e1e471e7f1ca62babe6f9b2a","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d425a9486551c717e1e471e7f1ca62babe6f9b2a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-13 14:34:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-13 14:35:09.000000000","tz":180},"subject":"Do not collect gnocchi project ID","message":"Do not collect gnocchi project ID\n\nCeilometer middleware now accepts also project names,\nso it\u0027s not required to gain project UUID [1].\n\nWe also switching gnocchi project name to `service` for all scenarios\nas it\u0027s already whitelisted by default [2]\n\n[1] https://opendev.org/openstack/ceilometermiddleware/commit/e2bf485044d8f3743da9298a9e461c5808be31f3\n[2] https://opendev.org/openstack/ceilometermiddleware/src/commit/082dabb1b7186f614664f69db32f80918b20d624/ceilometermiddleware/swift.py#L158\n\nCloses-Bug: #1879192\nChange-Id: I40d3178d1b3bd1234ac34c6fa1d0d2bcb7160720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/37fc6b9160856c27d9189c769297e2f9358bccf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/37fc6b9160856c27d9189c769297e2f9358bccf7"}]},"branch":"refs/heads/master"},"2ce00d6c69854a175d876c95b3439863a20a5fe4":{"kind":"REWORK","_number":3,"created":"2021-05-13 14:39:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/791250/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/791250/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/791250/3"}}},"commit":{"parents":[{"commit":"d425a9486551c717e1e471e7f1ca62babe6f9b2a","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d425a9486551c717e1e471e7f1ca62babe6f9b2a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-05-13 14:34:08.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-05-13 14:39:43.000000000","tz":0},"subject":"Do not collect gnocchi project ID","message":"Do not collect gnocchi project ID\n\nCeilometer middleware now accepts also project names,\nso it\u0027s not required to gain project UUID [1].\n\nWe also switching gnocchi project name to `service` for all scenarios\nas it\u0027s already whitelisted by default [2]\n\n[1] https://opendev.org/openstack/ceilometermiddleware/commit/e2bf485044d8f3743da9298a9e461c5808be31f3\n[2] https://opendev.org/openstack/ceilometermiddleware/src/commit/082dabb1b7186f614664f69db32f80918b20d624/ceilometermiddleware/swift.py#L158\n\nCloses-Bug: #1879192\nChange-Id: I40d3178d1b3bd1234ac34c6fa1d0d2bcb7160720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2ce00d6c69854a175d876c95b3439863a20a5fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2ce00d6c69854a175d876c95b3439863a20a5fe4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
