)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~806210","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2Fussuri~I5498030487677523ae71ae2f9a13074ab4552204","project":"openstack/openstack-ansible-os_swift","branch":"stable/ussuri","topic":"fix-fallocate-percent-stable/ussuri","hashtags":[],"change_id":"I5498030487677523ae71ae2f9a13074ab4552204","subject":"Revert \"split templates to work around configparser bug\"","status":"MERGED","created":"2021-08-26 15:30:45.000000000","updated":"2021-09-24 16:37:21.000000000","submitted":"2021-09-24 16:35:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806210-fix-fallocate-percent-stable/ussuri","cherry_pick_of_change":782956,"cherry_pick_of_patch_set":3,"meta_rev_id":"a8c0189e16781f090b22e2db235fd9bc7a4592da","_number":806210,"virtual_id_number":806210,"owner":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-24 16:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-09-24 14:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2021-08-26 15:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2021-09-24 14:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 15:30:45.000000000","updated_by":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-08-26 18:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-24 14:22:33.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"ec0c93f4e0589806efb2ac4033380e5223f6e8ba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"date":"2021-08-26 15:30:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* templates/account-server-2.conf.j2\n* templates/container-server-2.conf.j2\n* templates/object-server-2.conf.j2","accounts_in_message":[],"_revision_number":1},{"id":"acc0e392f7c5451e97e4bac64b7bfc66e08d082f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"date":"2021-08-26 15:32:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"efe0c0b5ce70c151baa99f584a54f2855b99941f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"date":"2021-08-26 15:33:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"de959865ffbef2cf41121b994437c35475de1e9b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-26 15:45:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1c8c5b1c93f8bf2a40d3acf477cd73358914b581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 18:15:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b70787711aea4c689688b1b8d838ac9f : SUCCESS in 10m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5a61041ec1c34250a61354bef27c9a4a : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9424bb5aeb264d538bf265bb5491c64f : SUCCESS in 1h 41m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/04bc248b93ea44f8819c9629445031e1 : SUCCESS in 1h 21m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f368f49cffd42a39b90bf1c8b921774 : SUCCESS in 1h 29m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bea4d9d63f94c6e93bb428b2c66f9b6 : SUCCESS in 1h 12m 24s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/49fad23e62e843bc87a02522a97c8604 : SUCCESS in 2h 34m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e2a25e3f7d741ada7e736675c9d302d : SUCCESS in 2h 09m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/38abfc4220144b1fa577143aa0d90878 : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1168c56ecb004f32b6969490dccd2985 : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29a24e0942dd46419bde49b3ad507286 : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/739b314884954bff895d067ae072f86e : SUCCESS in 57m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a154d4e2044977ac0443caa0256234 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ab33c7bd360f74440075ac83130540f88c631d97","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-09-24 14:22:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"29940548d753f3e43851b902c739636f4358066d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 14:30:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cd62bfae06ad6e169ed0fb9d9198200b99201629","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 16:35:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8a284c0928cb472db804517e034304d5 : SUCCESS in 14m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/08e2ebcedfbc4afeab2fc7bb0d037555 : SUCCESS in 1h 21m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8432f0a009c5481e972dd7a7c1943b34 : SUCCESS in 1h 50m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/14565aba65b44ba3ae056d8f5708e97b : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d99559a676644281bc8cb6131fbef412 : SUCCESS in 1h 22m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a29d6d7e8d844a8a9d5bb330de99cab5 : SUCCESS in 1h 12m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6db6eb8573ab45fe9ac20f36db3e737b : SUCCESS in 1h 10m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/42f74162ac634f838467ce5f06dee389 : SUCCESS in 1h 10m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53f964bb4c30458f94bff14f84e9cf87 : SUCCESS in 1h 05m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77dac6c48d234513be2850420d1686ac : SUCCESS in 1h 07m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b511acedc92447569bb48b6a79263f46 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":2},{"id":"285687504ad456b56630117bf537dd26a0784989","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 16:35:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a8c0189e16781f090b22e2db235fd9bc7a4592da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 16:37:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83dae2e1738a47a18d9ff9dfb5381499 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc55b10b8771b603be90a304a2c1906a89c7bae2","revisions":{"b93070fe1b44fe8b2206f8bc083b682666111d05":{"kind":"REWORK","_number":1,"created":"2021-08-26 15:30:45.000000000","uploader":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"ref":"refs/changes/10/806210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/10/806210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/10/806210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/10/806210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/10/806210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/10/806210/1"}}},"commit":{"parents":[{"commit":"d134e6849165578301ef74611a8b6b47ff97a479","subject":"Delegate gnocchi retrievement task to setup host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d134e6849165578301ef74611a8b6b47ff97a479"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-25 10:08:57.000000000","tz":0},"committer":{"name":"Adrien Cunin","email":"adrien@adriencunin.fr","date":"2021-08-26 15:30:45.000000000","tz":0},"subject":"Revert \"split templates to work around configparser bug\"","message":"Revert \"split templates to work around configparser bug\"\n\nThis reverts commit 3e151d97adbff78be1b146938374a0dced5970e9.\n\nReason for revert: Upstream bug has been fixed\n\nChange-Id: I5498030487677523ae71ae2f9a13074ab4552204\nRelated-Bug: #1872553\nCloses-Bug: #1921354\n(cherry picked from commit 24bbb72ac97bbd2683d0eae04b8ab4c3674005bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/b93070fe1b44fe8b2206f8bc083b682666111d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/b93070fe1b44fe8b2206f8bc083b682666111d05"}]},"branch":"refs/heads/stable/ussuri"},"fc55b10b8771b603be90a304a2c1906a89c7bae2":{"kind":"REWORK","_number":2,"created":"2021-08-26 15:32:17.000000000","uploader":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"ref":"refs/changes/10/806210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/10/806210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/10/806210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/10/806210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/10/806210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/10/806210/2"}}},"commit":{"parents":[{"commit":"d134e6849165578301ef74611a8b6b47ff97a479","subject":"Delegate gnocchi retrievement task to setup host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d134e6849165578301ef74611a8b6b47ff97a479"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-25 10:08:57.000000000","tz":0},"committer":{"name":"Adrien Cunin","email":"adrien@adriencunin.fr","date":"2021-08-26 15:31:44.000000000","tz":0},"subject":"Revert \"split templates to work around configparser bug\"","message":"Revert \"split templates to work around configparser bug\"\n\nThis reverts commit 3e151d97adbff78be1b146938374a0dced5970e9.\n\nReason for revert: Upstream bug has been fixed\n\nChange-Id: I5498030487677523ae71ae2f9a13074ab4552204\nRelated-Bug: #1872553\nCloses-Bug: #1921354\n(cherry picked from commit 24bbb72ac97bbd2683d0eae04b8ab4c3674005bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fc55b10b8771b603be90a304a2c1906a89c7bae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fc55b10b8771b603be90a304a2c1906a89c7bae2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
