)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~843289","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~If0e686352e08379027508c2939fae00db6ae6cb8","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"bug/1973045","attention_set":{},"removed_from_attention_set":{"34921":{"account":{"_account_id":34921,"name":"Danny Nelhams","email":"danny.nelhams@springernature.com","username":"dannynelhams"},"last_update":"2022-06-02 06:20:30.000000000","reason":"Danny Nelhams replied on the change"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-06-04 06:43:18.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"If0e686352e08379027508c2939fae00db6ae6cb8","subject":"Sync rings to all swift hosts to build_files","status":"MERGED","created":"2022-05-25 14:04:50.000000000","updated":"2022-06-04 11:02:50.000000000","submitted":"2022-06-04 11:01:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843289-bug/1973045","meta_rev_id":"a1be942f4ad6cdb0f947da9cd0eec14889ecff71","_number":843289,"virtual_id_number":843289,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-04 11:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-06-03 19:56:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-03 17:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2022-06-03 19:56:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"date":"2022-06-04 06:43:18.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2022-06-03 15:35:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":34921,"name":"Danny Nelhams","email":"danny.nelhams@springernature.com","username":"dannynelhams"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-25 16:21:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-26 02:14:33.000000000","updated_by":{"_account_id":34921,"name":"Danny Nelhams","email":"danny.nelhams@springernature.com","username":"dannynelhams"},"reviewer":{"_account_id":34921,"name":"Danny Nelhams","email":"danny.nelhams@springernature.com","username":"dannynelhams"},"state":"CC"},{"updated":"2022-06-03 17:00:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-06-03 19:56:31.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"21ddc78bfb2b51adcb01a0bbddc20b88fc8fd80d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-25 14:04:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8702b343989bf600d3c1f451d8a035fa2f416c95","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-25 14:13:03.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"33311382710fd2d01b989ef8dc0eb0fc4327fc6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 16:21:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9256d6d64c6149448534961df2ad9380 : SUCCESS in 14m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f60d5b4d48f04325830cd4d0980eed9e : SUCCESS in 1h 17m 03s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/de5a60da3b25485f9f58b0d47f3be4b3 : SUCCESS in 1h 30m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4294f07376d54c35a187a31a5fc066e9 : SUCCESS in 1h 35m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/c49857b997fb4dea89265936cadfefb3 : SUCCESS in 1h 36m 18s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/93da74970b1f45399cc0520a846ee3bf : SUCCESS in 2h 09m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddef0e9d978e4716b09155b971bb55f5 : SUCCESS in 2h 04m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd12467c1d62460f8f43ed7b596cf3bd : SUCCESS in 1h 17m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a42c63ac62848d2add81d97bbbd1e9e : SUCCESS in 1h 00m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f9fa0330d654ee090b0519a51d57378 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fb7d7295f8aa806b70bfcd191a5c13804503984d","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-25 18:52:17.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"8c175e08c1fe289d5f97304b9e21e21077ea92d9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-25 18:53:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"629283386274df7f52d2ba7ebbbbc84c89efe836","author":{"_account_id":34921,"name":"Danny Nelhams","email":"danny.nelhams@springernature.com","username":"dannynelhams"},"date":"2022-05-26 02:14:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dae5d3a5cc775be5666124bac07a3ebf1b62b007","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-26 06:51:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b7956b2d59ab1213be6422b02fae961123b28b0","author":{"_account_id":34921,"name":"Danny Nelhams","email":"danny.nelhams@springernature.com","username":"dannynelhams"},"date":"2022-06-02 06:20:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8bbf1a93256918166a899f50fc7d4ba6bd533fc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-03 15:35:12.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"e93fb944668f009b7ca304f23ee120da491b0357","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-03 17:00:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6ca9b4273e328b26e634b96f5efc4ea1292a037a","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-06-03 19:56:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1cf5083ed66f49e16de5c8c7fe7a5b018667e325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 19:58:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32e4d77c729934c2439ab0f53f7428c0506bd2dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 21:52:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e21782d70f6b406bb4deef5996966852 : SUCCESS in 19m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/52838bd98f474163af5b1f9466d3276b : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e4a2c08fd2a47aabd63264069690031 : SUCCESS in 1h 28m 53s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/04d3aeb90b174e55a9017104988bd264 : SUCCESS in 1h 29m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bc567fe93954d6699f98666a5a93866 : SUCCESS in 1h 33m 16s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/5746237b324945aea090e2ff431cd266 : FAILURE in 38m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85edb0b928724e9faf2930d099316f6e : SUCCESS in 1h 17m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09adb8741eb144a98149b27633fdab29 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7511e9287e1a2a347e7ed25774542d4a191f7733","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-04 06:43:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79a76fd549eb61b13ad253f6173dcf7faeaa7bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 09:07:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49664638948c4cf0919629db2d8d4d1d : SUCCESS in 12m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7d475f2d777744e0a0f2f56a7fa92ba1 : SUCCESS in 1h 20m 04s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd44e009dff14ec88d74834ea7ffb4b9 : SUCCESS in 1h 39m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e353884a410140b7a3097076719e7f6f : SUCCESS in 1h 27m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2c2ab186903478f9c4eacd43de7c4fe : SUCCESS in 1h 30m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3ba70e255454b8fb8f1eb24a83cc467 : SUCCESS in 1h 42m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3a794f024ca94118a66152ea6e22c600 : SUCCESS in 1h 48m 29s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e84fc2ad9eba4705932031ea86debdb7 : SUCCESS in 2h 16m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3ec45f7851a4b5fac42ac0d806b8f19 : SUCCESS in 2h 14m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/375f2bda34554396a663ea800d8d839d : FAILURE in 1h 16m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b7a0d46045d47248eac896a32e1391d : SUCCESS in 1h 20m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf21e6e4e1de4112a96f4503940f473f : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f952d960aa9b41a8eab6a704b74c851cc3dcbe3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 09:10:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b275c03b86f6b778a81d2ac420161164368115f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 11:01:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a779165be2f4109830b3067787dfc78 : SUCCESS in 13m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/796716ead8f04ee386ea36959a936a69 : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a6a43728c634039936eb4cb1a91f4b1 : SUCCESS in 1h 11m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c430c1f965d94713bb982fdad84ae2b6 : SUCCESS in 1h 05m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fb60ee5f06c4865b255b285a3272e1d : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/7df402978be0493fa52477c79e71dd93 : SUCCESS in 1h 49m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1af044d4b7134722a745e91e3837a925 : SUCCESS in 49m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd480ca04a74443ca6f3f7483808779a : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5cb1f9db90db0e5ff372a9e2db4d83d96012b355","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 11:01:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a1be942f4ad6cdb0f947da9cd0eec14889ecff71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 11:02:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3dfbf245e014761b8862faf7d93dd0b : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b02afd3ce1ddff3be8e62e88dd7528541cbebca","revisions":{"7b02afd3ce1ddff3be8e62e88dd7528541cbebca":{"kind":"REWORK","_number":1,"created":"2022-05-25 14:04:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/843289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/89/843289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/89/843289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/89/843289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/89/843289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/89/843289/1"}}},"commit":{"parents":[{"commit":"358f4a3aa896bec035947921a364addbcdc96a71","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/358f4a3aa896bec035947921a364addbcdc96a71"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-25 14:04:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-05-25 14:04:40.000000000","tz":120},"subject":"Sync rings to all swift hosts to build_files","message":"Sync rings to all swift hosts to build_files\n\nIn previous patch [1] I wrongly assumed that there\u0027s no reason to have\n/etc/swift/ring_build_files/\non all hosts. However, there\u0027s another logic in the role, that assumes\nhaving these files and verifying md5sum of them.\n\nAs an easy bugfix, we\u0027re returning sync to ring_build_files until we\nchange logic. It\u0027s also easily backportable.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_swift/+/765354\n\nChange-Id: If0e686352e08379027508c2939fae00db6ae6cb8\nCloses-Bug: #1973045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/7b02afd3ce1ddff3be8e62e88dd7528541cbebca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/7b02afd3ce1ddff3be8e62e88dd7528541cbebca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
