)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~888672","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I6bbb97cd2f1daac42269a335014eb6cf7f28d24e","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"osa/core-2.15","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-14 10:36:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6bbb97cd2f1daac42269a335014eb6cf7f28d24e","subject":"Fix linters and metadata","status":"MERGED","created":"2023-07-17 09:45:28.000000000","updated":"2023-08-14 10:37:25.000000000","submitted":"2023-08-14 10:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888672-osa/core-2.15","meta_rev_id":"fbc9c031591052db90c9724875b6bf50f738205c","_number":888672,"virtual_id_number":888672,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-14 10:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-08-08 21:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-14 08:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-14 08:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-17 12:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-08 21:00:56.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-08-14 08:43:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"ebd6e2f98d98d0d37b54f8af744652b0faae195e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-17 09:45:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffb6fdbdd3587853fd3fd93fd3e9d7c61ab98f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:10:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd4e418d6a854d8da9051521fc267d90\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5450f7b1556c400ea50474e15fbfb7d4 : SUCCESS in 10m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2874bc8d314e47669591878d4ac16a08 : SUCCESS in 1h 37m 12s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d44a708a759e462995938f246ed16c80 : SUCCESS in 1h 47m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1fdd568669c44569bc209b7f364d8e3e : FAILURE in 1h 17m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73fe318fad1e4f2dbdc807be603a8d5a : SUCCESS in 1h 47m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b1c10dc68ec42cfbee7ed0e06b0c02f : SUCCESS in 1h 31m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c1bfec7b13f44d6f94449052c9a91bf3 : FAILURE in 1h 01m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31d4a69c1e66433797b3ac1cb7ba6b72 : SUCCESS in 1h 34m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d216cc19f27426ba83a7a256551933c : FAILURE in 1h 05m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dda2fc7fb53c4187b7facb78225bd5bd : SUCCESS in 2h 21m 45s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8b808fc58464384af50539e947dc88b : FAILURE in 35m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa0078e055104d008b134c61f2ea829f : SUCCESS in 57m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c06fa2ab1c41a8877ab58bae30586d : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"50904652d1bc4a2e7214bfd3ebd51bc7dd464faa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-19 07:18:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd3e28537d2e95f6b7699efc72678e44d0f9ca53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 10:33:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3952fd8b1b044779a9daac406b598cd8\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/137a6a93025e4341b7be62117b55cd6f : SUCCESS in 16m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/75b59a54738340d98b7da79fa9a52599 : SUCCESS in 1h 46m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b57e87da85f4c6a98239d40cbea1ca6 : SUCCESS in 1h 46m 20s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/85c4bfbfe07646cbb0933e9efdffe657 : SUCCESS in 2h 05m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9670a217fbaa463a8c4e78cd85bca0fe : SUCCESS in 1h 49m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4a2d66f8bd742fe934cbfd3a753a28c : SUCCESS in 1h 52m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b369f191c4e48459f26cc7c40e4b7b4 : SUCCESS in 1h 59m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ad49b2650ad417bbcf7f92fb2d0c4ab : SUCCESS in 1h 47m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c65c6bdb6aa347cdbd182286ac8815f2 : SUCCESS in 2h 11m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0e4b63981e34af99cb9a50e27c1a19d : SUCCESS in 3h 02m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f9f871426ab4e4a9be8566928ec18ef : SUCCESS in 1h 10m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cc6cad9ad6e4fdfbb1d78f8ba8fa1b7 : SUCCESS in 1h 04m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b195f6b1b6ad4eeba3f34570b50280cc : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"61a550d96299e90b137f55d5c01c9d8f90e52418","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-08-08 21:00:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b24dd53232b64f0a2367d8cab2f3f9e6651aaeb6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-08-14 08:43:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90eef361fb217898ca9108eebc1805ca0a67e2f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 08:45:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa8540ced8e2e58891da1bd995c59f52bb38bfdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 10:36:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6a65d637890482a8a42ac1917547501\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49559a5d47044ed186f9b2aa0217d047 : SUCCESS in 14m 50s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a5a65e7d4ef34cd39d6ec761ff2ce98d : SUCCESS in 1h 37m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c00c585417434d6aaddc1a4eede1f8ac : SUCCESS in 1h 40m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4831f4d9f1b849048194fd0ae080166b : SUCCESS in 1h 42m 05s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2fddaf049f74243b906f5b1e46abb21 : SUCCESS in 1h 39m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09bfa27c8aca4ba891be7f37c03ca7b6 : SUCCESS in 1h 27m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d57daf3205d4037bb193a28c11844eb : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"21774c4c9354b55b2e59a7b2922274710b0ebb93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 10:36:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fbc9c031591052db90c9724875b6bf50f738205c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 10:37:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452c997dbbe6458bb82104da0538a73b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55798f602b604f3880d955e8da9c9e34 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99c6fb28355553e1d42ed1dfc8ec2e98fe077f8c","revisions":{"99c6fb28355553e1d42ed1dfc8ec2e98fe077f8c":{"kind":"REWORK","_number":1,"created":"2023-07-17 09:45:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/888672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/72/888672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/72/888672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/72/888672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/72/888672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/72/888672/1"}}},"commit":{"parents":[{"commit":"96a262b26b93cd0ccdaadf9263039b382d266da1","subject":"Add TLS support to swift backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/96a262b26b93cd0ccdaadf9263039b382d266da1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-17 09:45:20.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-17 09:45:20.000000000","tz":120},"subject":"Fix linters and metadata","message":"Fix linters and metadata\n\nWith update of ansible-lint to version \u003e\u003d6.0.0 a lot of new\nlinters were added, that enabled by default. In order to comply\nwith linter rules we\u0027re applying changes to the role.\n\nWith that we also update metdata to reflect current state.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/888223\nChange-Id: I6bbb97cd2f1daac42269a335014eb6cf7f28d24e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/99c6fb28355553e1d42ed1dfc8ec2e98fe077f8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/99c6fb28355553e1d42ed1dfc8ec2e98fe077f8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
