)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~892097","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2F2023.1~I4d6eacae2041b0a00114dda4e8315d4ec6295319","project":"openstack/openstack-ansible-os_swift","branch":"stable/2023.1","topic":"cherrypick-osa/member_role-nokmt6f95v","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-22 14:20:36.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I4d6eacae2041b0a00114dda4e8315d4ec6295319","subject":"Stop reffering _member_ role","status":"MERGED","created":"2023-08-21 13:14:09.000000000","updated":"2023-08-22 16:18:19.000000000","submitted":"2023-08-22 16:17:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892097-cherrypick-osa/member_role-nokmt6f95v","cherry_pick_of_change":891465,"cherry_pick_of_patch_set":1,"meta_rev_id":"e1bab0029eab4eec95c27eb284ab08ad9dfc235b","_number":892097,"virtual_id_number":892097,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 16:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-08-22 14:07:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-08-22 14:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2023-08-22 14:20:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-21 16:01:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-22 14:07:54.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"45f48ab6145095761f38842d6fc6ec543cdc2877","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 13:14:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bf90cc59ee321421fba0c384b23696a93276320e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 16:01:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c984946b8e854fb28b3be358bbe8797f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ef477d5123048429b54f66317c833a1 : SUCCESS in 14m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0fd1a2d93d674ea4bb4d245c40cc2cc7 : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/019bdfd7a44b453bac101921fa015091 : SUCCESS in 1h 50m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8a2f6c372b61406ba63c4dc84296cb12 : SUCCESS in 1h 58m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86ce176c25cb4cb08b203acb6b6a97ae : SUCCESS in 1h 58m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3e4788417984968b5da898c84687080 : SUCCESS in 1h 34m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28e1e07f17724a4f8fcac0f2852ad8bd : SUCCESS in 1h 50m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d8c3d4e0c7643d8ab70bd96c26a296f : SUCCESS in 1h 41m 14s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f89a251a8c7b4522b775dd6833d41563 : DISK_FULL in 2h 39m 14s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb9c1382bbef4489b6d76436813c70f2 : SUCCESS in 2h 09m 59s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1ad2a9ec0a24d958c80314f3915fd4d : SUCCESS in 2h 20m 05s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb61bc950d08412f82141ddc1ddcaa79 : SUCCESS in 1h 06m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f83d2fe1411d4a15b74c931292bddb14 : SUCCESS in 1h 07m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/022500a731cf45f79866e8c5b7a706ae : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e831039c1aa2275ce005943ad59030ec48b7f098","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 19:52:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cdf0614ce8b5ef5e0451af5241ba180799c13a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 22:33:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96e896a875b342809fd63c138be3020d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5a8e88393704007a99d9b7145398e2e : SUCCESS in 13m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/25af043148f14d8e9581047d25388455 : SUCCESS in 1h 13m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/31b93482662242a4823665699f8a0951 : SUCCESS in 1h 42m 31s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4b4b3a32b1f4fde9e0943dd25c4cd4a : SUCCESS in 1h 42m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/329a80048e2949798ebd7d8c65db63ba : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/397e24ee92d242d7a70ae771431a9bc4 : SUCCESS in 1h 46m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/68cdc5104d1843f58b07964f686abf70 : SUCCESS in 1h 44m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efc099219474460d926b600c4ea41c69 : SUCCESS in 1h 30m 45s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0800e27d5deb428bb53496732447085a : DISK_FULL in 2h 30m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/517fb049725f44ae8df4399d9f503772 : SUCCESS in 2h 03m 27s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad8d96150bb54f2eaad904b428207980 : SUCCESS in 2h 31m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4fb7ef634f404e5b8c65ca803b62f2e9 : SUCCESS in 1h 09m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8517d2870df4b8fa4ac08abe59ff59a : SUCCESS in 1h 06m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bbbecb482584738bb7663f9f5389c4a : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"09f23817b672f97061ce5ffc0a22a31712de2481","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 07:55:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d07a06fcdd9b5fcec8c6693f8e7a20f3a6d8182c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 10:27:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9719e484da4147bb8d86e881b7acc715\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8a9996647ee4574a390a1202084f7ea : SUCCESS in 16m 42s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/72ec17d434064eaea2130e92074c629f : SUCCESS in 1h 37m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a24b2a5b880e476dbc66e10d52c95f46 : SUCCESS in 1h 50m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f499219901014a05a9a637587b1d9dd5 : SUCCESS in 1h 48m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2bb07d287bf4cc4b6b4709c79dce757 : SUCCESS in 1h 55m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc01ca4a98714f5eb003a1010644d2df : SUCCESS in 1h 30m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8f6ea3a56c6e4b9c8a60e182ad5cd937 : SUCCESS in 1h 52m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfe1b527d12f44cfb2be1dc87e5ba4e1 : SUCCESS in 1h 24m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d0e34da38cf04da1bd84f6e5be9833e8 : SUCCESS in 2h 25m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ee38846fbeb4d549e0778a65eb0a0f0 : SUCCESS in 2h 25m 42s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01acaa25b6f4469d941adabdab6a866f : SUCCESS in 2h 14m 50s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/01dbe2f2dd2c4ab08068a66e7d55795c : SUCCESS in 1h 14m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4125c14622fb4fffa5995e0aeca2100c : SUCCESS in 1h 12m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6fd2ce8a524765a290d585324da676 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"44cf97cc252da1154379b076b4a3945e7792c06c","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-08-22 14:07:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24bc7426b9adde99994e4c09326384913bf2af50","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 14:20:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7801798336d909576f98b56f86b479d202a11bad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:22:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b17167f35af9ba07ad5c447fb4bf90d77d47f2d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 16:17:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78f04754738a46808b02b820168735c1\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad3b5e61c5b1411c8428d1523d06271d : SUCCESS in 15m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0c465aa10c7f47efaa2d94da92f7e1d2 : SUCCESS in 1h 27m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28e52f70159e4e2cbb91cc55482739cf : SUCCESS in 1h 49m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76d52dfd49704c6db116c0cfc8c9063e : SUCCESS in 1h 35m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb55c5ffda024232b2b699df5003426b : SUCCESS in 1h 49m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7080733eaa94565b449cc44d56b28c5 : SUCCESS in 1h 41m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa001f67bead47629edf05c0921ac7ed : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"733ee42cd47771a98e4a5c7bc6a3f25e94d18582","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 16:17:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e1bab0029eab4eec95c27eb284ab08ad9dfc235b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 16:18:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21ea87f6132a4fb092ea94d92aecc9ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1f697af22746c781ddc6c2a132dbd0 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"759957d1387a3fcdf224ab7a638cc1880a13cc56","revisions":{"759957d1387a3fcdf224ab7a638cc1880a13cc56":{"kind":"REWORK","_number":1,"created":"2023-08-21 13:14:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/892097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/97/892097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/97/892097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/97/892097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/97/892097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/97/892097/1"}}},"commit":{"parents":[{"commit":"eb526ea943ecd0685def7d06742c6263be75366e","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/eb526ea943ecd0685def7d06742c6263be75366e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-15 11:10:38.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-21 13:14:09.000000000","tz":0},"subject":"Stop reffering _member_ role","message":"Stop reffering _member_ role\n\nKeystone has stopped providing or reffering `_member_` role for a while,\nthus role should not be refferenced anymore.\n\nMoreover, with 2023.1 service policies have dropped `_member_`\nwhich resulted in the role to be insufficient for basic operations.\n\nChange-Id: I4d6eacae2041b0a00114dda4e8315d4ec6295319\nRelated-Bug: #2029486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/759957d1387a3fcdf224ab7a638cc1880a13cc56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/759957d1387a3fcdf224ab7a638cc1880a13cc56"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
