)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~946865","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Iacac451c556282888b4abe48e9e08afa80a745c9","project":"openstack/openstack-ansible-os_swift","branch":"master","attention_set":{},"removed_from_attention_set":{"27316":{"account":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"last_update":"2025-04-11 11:21:10.000000000","reason":"Change was abandoned"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-04-11 11:21:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iacac451c556282888b4abe48e9e08afa80a745c9","subject":"Migrate role to use SSH CA","status":"ABANDONED","created":"2025-04-10 05:59:55.000000000","updated":"2025-04-11 11:21:10.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9b74baacf513b4631078eebc47ddf91563a56568","_number":946865,"virtual_id_number":946865,"owner":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-04-10 15:46:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-10 08:08:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-10 08:52:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"9bbbcbf9c92139473472310c95e00350ef131147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 05:59:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c792c665e2d41836e59d9d0a28742c07743fb3b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 08:08:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d705ebdc4e140beb555f26390af4d8a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4e735d507b343f180831f43cf0e13bf : FAILURE in 6m 16s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6fcbedef87c54a2eb941efa1010597c6 : FAILURE in 1h 07m 39s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/88752ddaaad54adda5dbf8e399585856 : FAILURE in 1h 08m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a93b31a5dbb142398d920b9947bb17c3 : FAILURE in 49m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6f169fdd73e498d959ccc7a5d95b328 : FAILURE in 1h 11m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d97da0c9e9fa448595b4276abb0e433b : FAILURE in 57m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db9fb22e5f3c4b8f9e3479ba14d70057 : FAILURE in 55m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bce2e37af1894fff84ad9d0277a0949c : FAILURE in 1h 01m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c0b38f598724978bcc0cf0ad8f3dd65 : FAILURE in 1h 08m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7d0a1ae853340578ef0e4666b544f40 : FAILURE in 47m 42s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9e2ce4bbf4904b15a83f8b50b584fd0f : FAILURE in 55m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1dc8a77a19a42959fdd6e072102123e : FAILURE in 1h 54m 00s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc54bab2d95e44809f6395b68670a4c9 : FAILURE in 1h 51m 47s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98bb70ec231f4cada45ee03491ca4c68 : FAILURE in 48m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ded8c7c922747a6b437c8cb9d6af7a3 : FAILURE in 1h 47m 26s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/746ce8476d4545a3b75b84e59c31b733 : FAILURE in 32m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d7ed11a6e034488a0c9507d77c7e438 : FAILURE in 30m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9dd8f29252243ca8023cee186c3696b : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b1131629b79b45360e63ddfbd800e3b8a3920db7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 08:43:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1bdf9f536836bf3b1d61c8576b51a92d85844bff","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-04-10 08:52:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a6b9f630cab90137a8112ad3b2eb45acaa68b258","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 08:59:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0c1fcb60cce89ff2f6c25f29d0d6b4f796a5f87e","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 09:00:16.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8ae87ed5259729744d51e7c9ef9a3c3304f94995","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 10:41:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9c5cbfb0345265dec2fff2ac4b5f90812a56b954","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 10:56:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4d8962fe57168b31fd382475889a97264a601501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 11:46:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4986f69f53affab677efb6b3a4ea906896088e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 13:07:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dbb00a33b8a08c0d8895a6afd37712f35cc79388","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 13:26:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8cac299b87370012ecdbad0ea33b9e5840decc21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 13:33:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a62d71f9c685309ec1402d2b91d65d56a3fe311e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-10 13:39:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e6b39fc7f8b59092e5743864dcfc469bf599106b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 15:46:53.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd7a7dd9d62b428a918d953420559bda\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f655e385d0c4376b2b5bb2bf0f46962 : POST_FAILURE in 6m 49s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ce58ea7a757d472a98352a6143b3f974 : FAILURE in 1h 08m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/292192aa346048cc8c10f883a8419ed0 : FAILURE in 31m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/475572b2b104450ba55b51feef971e30 : FAILURE in 43m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a53120a545b402fbe43258aa919855c : FAILURE in 59m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e37e9e2aaa0d48f48771f59d122a8078 : FAILURE in 34m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/811691a0bd39416a9a18e68f852deece : FAILURE in 37m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/671f911699b5426a95de46b3b2d1d3ca : FAILURE in 57m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b5efd98f97543c8bba8e3fb3b335d0d : FAILURE in 1h 12m 37s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3e40c9b74c144fcf8b7d6ee095679836 : SUCCESS in 1h 59m 24s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8190b320bd6a4993ad10ff5ae1c8500c : SUCCESS in 1h 07m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9c3bba17ee54cac801b29ff1fdb5672 : SUCCESS in 2h 00m 37s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ca36af57fc24e32ad8cd67c5ee8faa3 : SUCCESS in 1h 46m 25s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/673b05acf3d64308b12f277b17af070c : SUCCESS in 1h 56m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0527875e9994be2a1373a7da0ccada5 : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/42aba4100f384bac9ca4b098abe86319 : FAILURE in 34m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30b90612fb0b41f4a4ce238831e4d8b6 : FAILURE in 48m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ebf71af89ba47ba97e10650acb61179 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":10},{"id":"9b74baacf513b4631078eebc47ddf91563a56568","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-11 11:21:10.000000000","message":"Abandoned\n\nThis change will be reintroduced in smaller chunks to avoid ongoing feature creep","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"c5a783b44df3804dd5aad61688f023d567b0e1a7","revisions":{"dc8df120b57105fc6639faca818c734a343479ed":{"kind":"REWORK","_number":1,"created":"2025-04-10 05:59:55.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/1"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/dc8df120b57105fc6639faca818c734a343479ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/dc8df120b57105fc6639faca818c734a343479ed"}]},"branch":"refs/heads/master"},"f0c46b6762c98a7f003d66c0240b5716dab92d0a":{"kind":"REWORK","_number":2,"created":"2025-04-10 08:43:20.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/2"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 08:43:16.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/f0c46b6762c98a7f003d66c0240b5716dab92d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/f0c46b6762c98a7f003d66c0240b5716dab92d0a"}]},"branch":"refs/heads/master"},"94fe467b8e96296c44641afde2b4b49aa5ad9810":{"kind":"REWORK","_number":3,"created":"2025-04-10 08:59:35.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/3"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 08:59:29.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/94fe467b8e96296c44641afde2b4b49aa5ad9810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/94fe467b8e96296c44641afde2b4b49aa5ad9810"}]},"branch":"refs/heads/master"},"9df4dbac7820167a0df47aabe8e1b3e1e358d0f0":{"kind":"REWORK","_number":4,"created":"2025-04-10 10:41:20.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/4"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 10:41:14.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/9df4dbac7820167a0df47aabe8e1b3e1e358d0f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/9df4dbac7820167a0df47aabe8e1b3e1e358d0f0"}]},"branch":"refs/heads/master"},"6032bd22c8c0c0d5bd2c35e03bbc5c45bc7d70c3":{"kind":"REWORK","_number":5,"created":"2025-04-10 10:56:55.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/5"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 10:56:51.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/6032bd22c8c0c0d5bd2c35e03bbc5c45bc7d70c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/6032bd22c8c0c0d5bd2c35e03bbc5c45bc7d70c3"}]},"branch":"refs/heads/master"},"b15c9eab8849baf3cb02b3251a6b56db6ad175df":{"kind":"REWORK","_number":6,"created":"2025-04-10 11:46:44.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/6"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 11:46:40.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/b15c9eab8849baf3cb02b3251a6b56db6ad175df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/b15c9eab8849baf3cb02b3251a6b56db6ad175df"}]},"branch":"refs/heads/master"},"7af2df95ebd672b0eb071a4d461200096080866a":{"kind":"REWORK","_number":7,"created":"2025-04-10 13:07:23.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/7"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 13:07:19.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/7af2df95ebd672b0eb071a4d461200096080866a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/7af2df95ebd672b0eb071a4d461200096080866a"}]},"branch":"refs/heads/master"},"868070e04dcb0f8cc65778bd6b32aa531a91b9f6":{"kind":"REWORK","_number":8,"created":"2025-04-10 13:26:51.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/8"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 13:26:45.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/868070e04dcb0f8cc65778bd6b32aa531a91b9f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/868070e04dcb0f8cc65778bd6b32aa531a91b9f6"}]},"branch":"refs/heads/master"},"fe420b32d13ce4d8920679f04606ce0c40eb4a0e":{"kind":"REWORK","_number":9,"created":"2025-04-10 13:33:14.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/9"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 13:33:10.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe420b32d13ce4d8920679f04606ce0c40eb4a0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe420b32d13ce4d8920679f04606ce0c40eb4a0e"}]},"branch":"refs/heads/master"},"c5a783b44df3804dd5aad61688f023d567b0e1a7":{"kind":"REWORK","_number":10,"created":"2025-04-10 13:39:31.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/65/946865/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/65/946865/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/65/946865/10"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 05:59:41.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-10 13:39:27.000000000","tz":0},"subject":"Migrate role to use SSH CA","message":"Migrate role to use SSH CA\n\nChange-Id: Iacac451c556282888b4abe48e9e08afa80a745c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c5a783b44df3804dd5aad61688f023d567b0e1a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c5a783b44df3804dd5aad61688f023d567b0e1a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
