)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~947050","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"refactoring","attention_set":{},"removed_from_attention_set":{"27316":{"account":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"last_update":"2025-05-14 13:02:34.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-14 06:56:47.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3","subject":"Remove seemingly unsused mlocate cronjob","status":"MERGED","created":"2025-04-12 06:22:27.000000000","updated":"2025-05-14 13:03:57.000000000","submitted":"2025-05-14 13:02:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947050-refactoring","meta_rev_id":"6a2198684ea2ba2ddd7678c6affc0785d5e963c9","_number":947050,"virtual_id_number":947050,"owner":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-14 13:02:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-05-14 06:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-14 11:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-14 11:24:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-12 07:47:53.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-04-14 18:35:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 11:24:11.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"67c8c26a6eeb7d8c6ca543f262f139669674a3b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-12 06:22:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59f60153dfb952aa52547a312120fdb95806c6b4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-12 07:47:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c414beab8dc1ef3c8bf725a070cc1cd66db22f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-12 07:55:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7c46b3afd7c8b8ebb4596177679950a19a56ddb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-12 08:31:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"59ace6ebcd0db07598586136e82751a01106a617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-14 15:47:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_27316\u003e","accounts_in_message":[{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"}],"_revision_number":4},{"id":"8a259e4ea2843ddd59c868991317ee8440e32cea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-14 15:48:30.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"08d11f29a446a3f770eb7d4ebb77ee99f0d598a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-14 15:57:00.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_27316\u003e","accounts_in_message":[{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"}],"_revision_number":6},{"id":"37ac65a120c43b02723e77a871951048607ff05f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-14 16:37:32.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"9c08b025f17a686aee1a9b2631aab33bf05b40b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 18:35:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d927bf82a062436b93d7b79eeb8d8c45\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04378b94216847328d702adc886901fd : SUCCESS in 6m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f63af49493a4459b12c15d4e991d9a8 : SUCCESS in 1h 08m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4763c61f1264b0da3d96257f202de95 : SUCCESS in 1h 09m 19s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d46958fde9664aab931e08a31106ba19 : SUCCESS in 1h 07m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a75c7cb2f59d4c489376c81bf75410d8 : FAILURE in 59m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44bc10aad13a4ecd97965c6222dad17a : SUCCESS in 1h 11m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/444eb8c74fc345269c4cee883cd59903 : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d5015d540c74d4aa93a86bdb7246206 : SUCCESS in 45m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77a03762c093484b915a4d4ce34b17ae : SUCCESS in 41m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e0de4fcd7abc430fbf7528b53cbf92e2 : SUCCESS in 1h 43m 43s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3cabeb129e814e40b67fb67759f209ce : SUCCESS in 1h 50m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8be709922846469db106079975dc5c43 : SUCCESS in 1h 55m 39s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76946ee6cc444416bf8d696d55020a3f : SUCCESS in 1h 06m 51s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b0b497824b24739986ea7f2450c69dc : SUCCESS in 1h 01m 28s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9b96d6860124710bd6d6470c98ada00 : FAILURE in 57m 43s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c04d56115ce94104b386342f396a82ad : FAILURE in 35m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff080830a69b49c4b59fcaa5daa59324 : SUCCESS in 1h 10m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0007d3967ca46b8aacff3d0bcff7e92 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":7},{"id":"83ef4a9a09e275b23965ffbc500882e28f6bb0eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-13 18:15:38.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_27316\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"}],"_revision_number":8},{"id":"a99cc9bd258f00390b2b2c48643072947fb08105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 20:34:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8317e29512104a6e81474e5b27a65054\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/859b95f01deb454a81b861a9fb75edf8 : SUCCESS in 6m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/21f73e88f41943e19e6753d08232f56c : SUCCESS in 42m 54s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/08806e9edd0741b8af25b533dd3d0184 : SUCCESS in 1h 27m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc4ef366ede24a4ba68e2221a0726fdd : SUCCESS in 1h 09m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/602cd28334d94264bc3538303080c369 : SUCCESS in 1h 25m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0763f6a549c34130b37e7a0c8d802505 : SUCCESS in 1h 34m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2b3fe206dcf478494075acd54d29e70 : SUCCESS in 1h 11m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe78eebed0be4099b205e666683e3513 : SUCCESS in 40m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81f24f324b56463c8f0e1e4c94126a3b : SUCCESS in 1h 27m 30s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e1a03502d1bb466082d72321ad0fcb83 : SUCCESS in 1h 08m 30s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/937cb063aaa54056a4452e92d87fe534 : SUCCESS in 2h 05m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f332d65bda35429ea93158800dc886f4 : SUCCESS in 2h 06m 22s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81259eea26324db0b0d40deed7952b5f : SUCCESS in 1h 52m 25s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9684ad2cf73c418aa3a411ef5df133e3 : SUCCESS in 2h 15m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/220c738f86324901b80cbbaf88e8b724 : SUCCESS in 2h 10m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/135afa396d204853a010016dcb97b7cb : SUCCESS in 1h 09m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e9022324153406186c217876f398ecf : SUCCESS in 43m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0635153bae42f0ab4cf5f61ffa6ec0 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":8},{"id":"879c58561dc924ab0d73cb32e886092bce1a3afd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-14 06:56:47.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"8efc9d624eea52aa60a497d89ff4b7576d082d22","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-14 11:24:11.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"aa164ca868eb1abfb2f9aa11a1b88772cddb1cc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 11:26:12.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"eddf3ffdcf64142657ae7fac9db07a893b7368a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 13:02:34.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b1f7ee0c321431685fbc9f34715f029\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddc5155e28a340759163a08d9c0ebd00 : SUCCESS in 4m 59s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c4a8225aa8ae49fda426b095c9d7bac6 : SUCCESS in 39m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9d39970350cb4b63b8a2d879564bce09 : SUCCESS in 1h 28m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78b16ae8b3004115b704447d9ffdffe3 : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/444a1f07d7f44f8893dd0ce7dbb116b3 : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/371f36046549437c9a283a9811538681 : SUCCESS in 1h 13m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5603901ea164137bc375ca9f920c519 : SUCCESS in 1h 28m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb26dc78d46a4944b6318c64aa988f83 : SUCCESS in 1h 34m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f0e0be8e25422ea0c15925576254df : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":8},{"id":"50c658ebfbd7637ea48a6a02e1fedf915a9571d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 13:02:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"6a2198684ea2ba2ddd7678c6affc0785d5e963c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 13:03:57.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7d2af869e694b8788eb90c58fc5c1e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b47e61b4384447f8999d66547570ec98 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"3e829c61211673e0a6da91500401e4440ec1f356","revisions":{"52d22ffdd074d88189eb9b0a719b4e6a3b936f9d":{"kind":"REWORK","_number":1,"created":"2025-04-12 06:22:27.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/50/947050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/947050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/1"}}},"commit":{"parents":[{"commit":"658aef2aea12004afbd6dd3a98e34645bb8b2f99","subject":"swift-object-expirer is its own distro package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/658aef2aea12004afbd6dd3a98e34645bb8b2f99"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 06:22:09.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 06:22:09.000000000","tz":0},"subject":"Remove seemingly unsused mlocate cronjob","message":"Remove seemingly unsused mlocate cronjob\n\nChange-Id: Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/52d22ffdd074d88189eb9b0a719b4e6a3b936f9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/52d22ffdd074d88189eb9b0a719b4e6a3b936f9d"}]},"branch":"refs/heads/master"},"cb55064a17f64b9f5f91e7a614b7c9298aba66cd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-12 07:55:22.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/50/947050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/947050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/2"}}},"commit":{"parents":[{"commit":"fe12436d2b9dcd3810a4daf2e2a314d66c1e743a","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/fe12436d2b9dcd3810a4daf2e2a314d66c1e743a"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 06:22:09.000000000","tz":0},"committer":{"name":"Daniel Preussker","email":"f0o@devilcode.org","date":"2025-04-12 07:55:22.000000000","tz":0},"subject":"Remove seemingly unsused mlocate cronjob","message":"Remove seemingly unsused mlocate cronjob\n\nChange-Id: Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/cb55064a17f64b9f5f91e7a614b7c9298aba66cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/cb55064a17f64b9f5f91e7a614b7c9298aba66cd"}]},"branch":"refs/heads/master","description":"Rebase"},"87831c902065d66447bbb2bea85203d36711d1ff":{"kind":"REWORK","_number":3,"created":"2025-04-12 08:31:20.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/50/947050/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/947050/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/3"}}},"commit":{"parents":[{"commit":"658aef2aea12004afbd6dd3a98e34645bb8b2f99","subject":"swift-object-expirer is its own distro package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/658aef2aea12004afbd6dd3a98e34645bb8b2f99"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 06:22:09.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 08:31:17.000000000","tz":0},"subject":"Remove seemingly unsused mlocate cronjob","message":"Remove seemingly unsused mlocate cronjob\n\nChange-Id: Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/87831c902065d66447bbb2bea85203d36711d1ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/87831c902065d66447bbb2bea85203d36711d1ff"}]},"branch":"refs/heads/master"},"419ca39a9b936fe8945662f8f4a8209a0ce759cc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-04-14 15:47:42.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/947050/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/947050/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/4"}}},"commit":{"parents":[{"commit":"bbd3383f56991ead60523da71816490089102503","subject":"swift-object-expirer is its own distro package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/bbd3383f56991ead60523da71816490089102503"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 06:22:09.000000000","tz":0},"committer":{"name":"Daniel Preussker","email":"f0o@devilcode.org","date":"2025-04-14 15:47:42.000000000","tz":0},"subject":"Remove seemingly unsused mlocate cronjob","message":"Remove seemingly unsused mlocate cronjob\n\nChange-Id: Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/419ca39a9b936fe8945662f8f4a8209a0ce759cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/419ca39a9b936fe8945662f8f4a8209a0ce759cc"}]},"branch":"refs/heads/master","description":"Rebase"},"27e02d1350464ea7506ec66379837db7a5757893":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-04-14 15:48:30.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/50/947050/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/947050/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/5"}}},"commit":{"parents":[{"commit":"843865196624dab69609392763fdb8f3d05b57f7","subject":"swift-object-expirer is its own distro package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/843865196624dab69609392763fdb8f3d05b57f7"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 06:22:09.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-14 15:48:24.000000000","tz":0},"subject":"Remove seemingly unsused mlocate cronjob","message":"Remove seemingly unsused mlocate cronjob\n\nChange-Id: Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/27e02d1350464ea7506ec66379837db7a5757893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/27e02d1350464ea7506ec66379837db7a5757893"}]},"branch":"refs/heads/master"},"c68cd1f7bd6f5a59653719b916e46c5cdeec0f70":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-04-14 15:57:00.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/947050/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/947050/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/6"}}},"commit":{"parents":[{"commit":"39f2e65433a0f530a81b079bf7847f2fe24ded7b","subject":"swift-object-expirer is its own distro package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/39f2e65433a0f530a81b079bf7847f2fe24ded7b"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 06:22:09.000000000","tz":0},"committer":{"name":"Daniel Preussker","email":"f0o@devilcode.org","date":"2025-04-14 15:57:00.000000000","tz":0},"subject":"Remove seemingly unsused mlocate cronjob","message":"Remove seemingly unsused mlocate cronjob\n\nChange-Id: Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c68cd1f7bd6f5a59653719b916e46c5cdeec0f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/c68cd1f7bd6f5a59653719b916e46c5cdeec0f70"}]},"branch":"refs/heads/master","description":"Rebase"},"2f0cf5d64b64a83c2488df4702940c3fbda16fbd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-04-14 16:37:32.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/50/947050/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/947050/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/7"}}},"commit":{"parents":[{"commit":"47bb2829fbf6fb399488fef747efdbe36f320072","subject":"Migrate cronjobs to systemd-timers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/47bb2829fbf6fb399488fef747efdbe36f320072"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 06:22:09.000000000","tz":0},"committer":{"name":"Daniel Preussker","email":"f0o@devilcode.org","date":"2025-04-14 16:37:32.000000000","tz":0},"subject":"Remove seemingly unsused mlocate cronjob","message":"Remove seemingly unsused mlocate cronjob\n\nChange-Id: Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2f0cf5d64b64a83c2488df4702940c3fbda16fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2f0cf5d64b64a83c2488df4702940c3fbda16fbd"}]},"branch":"refs/heads/master","description":"Rebase"},"3e829c61211673e0a6da91500401e4440ec1f356":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-05-13 18:15:38.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/947050/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/50/947050/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/50/947050/8"}}},"commit":{"parents":[{"commit":"d3b125cb66e337a8309a2e052d91bf0ca786ec21","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/d3b125cb66e337a8309a2e052d91bf0ca786ec21"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-12 06:22:09.000000000","tz":0},"committer":{"name":"Daniel Preussker","email":"f0o@devilcode.org","date":"2025-05-13 18:15:38.000000000","tz":0},"subject":"Remove seemingly unsused mlocate cronjob","message":"Remove seemingly unsused mlocate cronjob\n\nChange-Id: Ibd824e2286d477fe8294d0a84ea8b592cdad0fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3e829c61211673e0a6da91500401e4440ec1f356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3e829c61211673e0a6da91500401e4440ec1f356"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
