)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~948194","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I0c359a538b63c9e5f7ef77e5ca15537f3c509964","project":"openstack/openstack-ansible-os_swift","branch":"master","attention_set":{},"removed_from_attention_set":{"27316":{"account":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"last_update":"2025-05-06 14:59:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c359a538b63c9e5f7ef77e5ca15537f3c509964","subject":"Add Ceilometer middleware to distro packages","status":"MERGED","created":"2025-04-25 13:31:51.000000000","updated":"2025-05-06 15:00:53.000000000","submitted":"2025-05-06 14:59:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948194","meta_rev_id":"49e705373e6fc548a85104c71bf239c02f7ba442","_number":948194,"virtual_id_number":948194,"owner":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-06 14:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-04-25 13:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-02 07:34:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-02 07:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-25 13:32:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-04-25 16:05:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-02 07:34:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"878bec6e21449a60a45c73fdbdd562d6f5ad9dd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"date":"2025-04-25 13:31:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ed0cd383ef45d933a246d33ac6c8944a247a96e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-25 13:32:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1695811daa287880cc7c025c3f2beca4d9b913f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 16:05:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f08660f47ea48ca9e51ac33882f2e07\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adb77fb10052457e8fd97e9131cbfd32 : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/512a4656197b405cbbd69d97d4a2474e : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/571d40980f1e4029a180e2f18445eae1 : SUCCESS in 1h 17m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4f6d275ff7e494e99bcce4510a6031f : SUCCESS in 1h 16m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10c75dc32f094363b647cf5ee95981e5 : SUCCESS in 1h 24m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b47cf95f97a458eaf4687902c4c63ea : SUCCESS in 1h 13m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/17469bde945e49ed9a5b686ddd0514bd : SUCCESS in 1h 28m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f6dbc4f0b924157a713226b193f2157 : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3970712a6a7c4936911234e5b78fd7e1 : SUCCESS in 1h 36m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a898d3e694bf469abbfb3516f3365917 : SUCCESS in 1h 51m 42s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49969bdaadbd422b907db7f5700e75b9 : SUCCESS in 2h 27m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c899fcbddf24488ea724f9f2ca24cfcd : SUCCESS in 2h 11m 47s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5068cdbbe544cbeab67d6ad03278472 : SUCCESS in 2h 13m 12s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59d56acd7bd34c0b860a493e5c39d8e5 : SUCCESS in 2h 05m 00s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6313be47381a43f8be064d62b8e424a4 : SUCCESS in 2h 04m 33s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8cd950b5c5d74fcb8cec962ab30bdcb6 : FAILURE in 27m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/feae2860735c4c92aec73f35e77fcf36 : SUCCESS in 1h 02m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdfeb5510cde4903b8aeb8c5dd360860 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"dbe603193419a88155ac3a56546c70aba8002bd4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-02 07:34:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"744b65f231c06bb5f06d1da5fe305bfb221848b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 07:36:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fed3e40884660d14bd9abb3c45d087df2a861d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 09:12:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae078d08ac464004ab2372ff41ca82d5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d939d43ca184eecbc054f726c35db56 : SUCCESS in 6m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c3761d8943e349019454be13ae99340e : SUCCESS in 1h 30m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/530bb0c4f80a4cdfa0238853a255a783 : SUCCESS in 1h 23m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/064fbef318ce4456a811704393381d60 : SUCCESS in 1h 24m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb89c7f1f13449bd8a796d2a41fc1e45 : FAILURE in 43m 42s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28b0bbdaec694967af6b7ee48cf50ba2 : SUCCESS in 1h 10m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/244ffcf44c874a0d98131a4cd549fde6 : SUCCESS in 1h 24m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c71d8af7953e432ea2a9ad956c002cbf : SUCCESS in 1h 29m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ceadbca971b4ffda21922e29eae2074 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"050b1b3f55e74107bfaffbf32d77098e0a4b3e70","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-02 09:23:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3966bd2fd369cd2a83b3a9d825b37b29daed6ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 11:33:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20ff8eac26fc45c5b31ead034e5c47ea\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fc37ecde7e944408c63ccf30d646150 : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b528a60b776c4f409c92595103fa51db : SUCCESS in 1h 24m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaf47bd0ddb74e5991d373ca8710d0d9 : SUCCESS in 1h 19m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2fb60c347fc14ae1afd17bdc76e0e435 : SUCCESS in 1h 19m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc6c110448bc4191a33fdb6f94b4a607 : SUCCESS in 1h 15m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1811ca2e8cb432d90323dae936dfbd4 : SUCCESS in 1h 32m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cb2ab2bc7b764d2eb727f55e8bfef7cb : SUCCESS in 1h 32m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82e804d626c64c9c9ebe268489f99497 : SUCCESS in 1h 21m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3170499f76241fe8f7f44f4f8aa5c6a : SUCCESS in 42m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f9ac56609f6c4d9aa8797b120e13f3d5 : SUCCESS in 1h 01m 12s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/503b53c67c0e4c4d8cc91812b0a4ef0d : SUCCESS in 1h 00m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7eba9792c0c44ce2aca4fb8a9500ce17 : SUCCESS in 55m 38s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c440c939ef7841ea8b9f0ba41d2acdec : SUCCESS in 54m 47s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51f18d7324ab4f5484d9839e95fb14a7 : SUCCESS in 1h 38m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8e68e9ec60e48eaa96d1cf896940a34 : SUCCESS in 2h 07m 16s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3aed74663b7d46af8d9de6149efadadf : FAILURE in 25m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f991c24e937043a3a275e74450ac31e5 : SUCCESS in 38m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7847cb52bfa74dd99f37be2125ac93e2 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1c2287e2c8f2fbe2c53f3f6a64bf839302221875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 11:35:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9da15cf758cb22b746c142521f69207186d3ef9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 13:02:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f44f838db8f495f86d9bee9a465729b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d42db91a5aa4525944cf11745448402 : SUCCESS in 6m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/723b8e566cfc4655872df6f1a023f7c3 : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26dfa8b140df4f4db4ec823b09ac7a25 : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c413dc0fa2ad4d7fae540b42997a7997 : SUCCESS in 1h 18m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa3bf6a9e2064a3ebba66646cea6a354 : FAILURE in 31m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/792455ad9101466f80d0179b7f4d584f : SUCCESS in 1h 09m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f8d26f85f77408598392c910c67408f : SUCCESS in 1h 07m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d7aef18332648bfa46e6b5bfe6f7388 : SUCCESS in 1h 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc3aa345bda4c15bc656eec7cd5b6d1 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"498c0b1e2935d3e3b83164a489b239e004decaa9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-06 10:54:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef7ec3e33623e095cfa6bb36fd7597afdee00d74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 13:18:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc984577873f4a5f89a67c9b229d73b4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/563e4e1e7f3c4d68b1917851322b2725 : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ee0bd113ccbd4766af12e80d127b4d0c : SUCCESS in 1h 04m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/134bd68e5de54cc9bfb34d663ebafed7 : SUCCESS in 1h 25m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/adeda248e81a4957ae0aff69e620dd84 : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c543ad73852d4f70a8b3450132210f3a : SUCCESS in 1h 14m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34f2e7b6b5e149febb4db6ad25945fd7 : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d80828684e2248bdacc416e0798ac80d : SUCCESS in 1h 26m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a5fc2b4c89f43f999ef793d6c1e5d44 : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c8afdb955b34c7690c07519be95c3a1 : SUCCESS in 1h 29m 34s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/25dd4a409f3d450883c16d0016f4cdcf : SUCCESS in 2h 04m 35s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d48ab3ec09c34e54be9533cc0217f512 : SUCCESS in 2h 16m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e3e86302bd94c1cbc718e3729012161 : SUCCESS in 1h 36m 40s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf065649752749188b32f2dd9a68f199 : SUCCESS in 2h 01m 27s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/288717d6e7fb450e9f04d7eab6d5f44e : SUCCESS in 2h 09m 08s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeb29d2b70164b3584c6609c9fc68b5f : SUCCESS in 1h 39m 41s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/35a3bb4881524dc686c08ffe8aa87422 : SUCCESS in 1h 03m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83b4ad5b4f7549739f3b31e736e3536e : FAILURE in 45m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60f5bc69424e463182f88ea1a455ac37 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"647589b31abeada43fdc3f44bdcdaaaaefb46930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 13:20:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0cc9a757e5b5eb06fd7ef5d5e129dd3267f324a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 14:59:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f8f88f4e2054ff28000fbeb88f4d265\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4aa7b08c572481490a606ebe17a8d74 : SUCCESS in 3m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/12cd930e2f4b4f57af5ae469cc22bd65 : SUCCESS in 1h 02m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b7c0591af834195b9f66086eb366a38 : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e5d9db86df641269ab0eb6439f9c312 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b8bbc2b24a44af68a028a7f0ce41ffb : SUCCESS in 1h 34m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/935162243d8a4468bad90b28fd33c39c : SUCCESS in 1h 34m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cd4e835ce0f41c3aeeab450b7d2b36b : SUCCESS in 1h 30m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/692245f43f314453ac0876f22c4eedd4 : SUCCESS in 1h 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe232a0f46147a5a9fe42584ff34766 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"cb22b3d5b26d8251e6abea77ff1c90728cd4eb5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 14:59:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"49e705373e6fc548a85104c71bf239c02f7ba442","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 15:00:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f3ec23f7f34f72910025e42c0b6e73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46d17c817cf848089e40e9f41e2757d6 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b6decd1270fd21f2b762ceb8d5d9f9e72502cc1","revisions":{"9b6decd1270fd21f2b762ceb8d5d9f9e72502cc1":{"kind":"REWORK","_number":1,"created":"2025-04-25 13:31:51.000000000","uploader":{"_account_id":27316,"name":"Daniel Preussker","email":"openstack@devilcode.org","username":"f0o"},"ref":"refs/changes/94/948194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/94/948194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/94/948194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/94/948194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/94/948194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/94/948194/1"}}},"commit":{"parents":[{"commit":"47bb2829fbf6fb399488fef747efdbe36f320072","subject":"Migrate cronjobs to systemd-timers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/47bb2829fbf6fb399488fef747efdbe36f320072"}]}],"author":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-25 13:29:36.000000000","tz":0},"committer":{"name":"Daniel \u0027f0o\u0027 Preussker","email":"f0o@devilcode.org","date":"2025-04-25 13:31:31.000000000","tz":0},"subject":"Add Ceilometer middleware to distro packages","message":"Add Ceilometer middleware to distro packages\n\nChange-Id: I0c359a538b63c9e5f7ef77e5ca15537f3c509964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/9b6decd1270fd21f2b762ceb8d5d9f9e72502cc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/9b6decd1270fd21f2b762ceb8d5d9f9e72502cc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
