)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~977841","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I97fd4af8f70fa5468ea2b7e3213905d5d4b9d6a4","project":"openstack/openstack-ansible-os_swift","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-17 14:04:23.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-17 14:04:23.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-17 08:10:57.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I97fd4af8f70fa5468ea2b7e3213905d5d4b9d6a4","subject":"Run lxc jobs instead of metal","status":"MERGED","created":"2026-02-24 16:56:37.000000000","updated":"2026-04-17 14:09:14.000000000","submitted":"2026-04-17 14:04:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977841","meta_rev_id":"046a88744509baa755f5b90bad744eb594361eac","_number":977841,"virtual_id_number":977841,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-17 14:04:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-04-17 08:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-17 08:10:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-04-17 08:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2026-04-17 08:10:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 19:33:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-25 10:03:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-02-25 10:03:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-04-01 16:16:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-04-05 15:28:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-04-05 15:28:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-04-05 15:28:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"bc571656bfade6e1352d23c8e10719c8e98aa32e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-24 16:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4de1280239f7c63a1df5a01f65aa2b2ae9617ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 19:33:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/115ac4771e9b41e3ade95a9b41e622fc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c59d6a701ff146cfa7832b74638ead0c : SUCCESS in 6m 52s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/330dee47561947518eed4cca73006c94 : FAILURE in 1h 32m 30s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/daeb00285064412a89da45fd54b2d082 : SUCCESS in 1h 01m 34s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8a35f7d6a7d3445e9f595c26fd63db81 : FAILURE in 1h 02m 08s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9155a1c58c684640b14b02d36f7579a9 : SUCCESS in 1h 05m 25s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfbd675df9f041eeaf64443231d40cef : SUCCESS in 2h 31m 49s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e3df9e04714d4f74a4042d3ca550b130 : FAILURE in 7m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c53e271e9eaf41f6b6539788d83e6a9d : FAILURE in 45m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144dc7f2a6c44023b147cdedea4cc02c : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"448e49ebda0a1cc5811f0b2d8962d0eafb816bae","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-25 09:44:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c89ea01d535994364e9dd809189c30f33ec0d814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:12:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e2612aec10c447e802fce3caa6164a0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea0170648894499c9d1ffccf1b261617 : SUCCESS in 8m 30s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/6df909c0431146898ca56c283bc6e607 : FAILURE in 1h 30m 33s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ea7366e5cb0d4cad9fb57bb6027b9087 : SUCCESS in 1h 42m 43s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/62700752e2914aa4b41f3062288a6bec : SUCCESS in 1h 59m 41s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/741769649d9e49b0ab1748afb706aa9d : SUCCESS in 1h 53m 23s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86c86e5702f14ef0b0d2fc6b8268b66e : SUCCESS in 2h 21m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/986859a7e7924279a6dba985f9696a7d : FAILURE in 10m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4732cc5b556146d7b4b4f79d8597eed9 : FAILURE in 1h 01m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12152cb75ecf4823960986d659ae0513 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4b69c86431eea541c9df77b4b1f9b0d50cdac5b5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-09 08:34:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b171e20c11b24a7bd20937c755a527b805fe60c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 11:06:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c622d309bd4f408d86a1108bf0a404e2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f189a0b1bf843f39295e8af1d0b7c29 : SUCCESS in 13m 32s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/21984a632aa145fc81874b0251997953 : FAILURE in 1h 45m 47s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/09d2c63d327d44d48d5cd7553f4196c4 : FAILURE in 1h 56m 49s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/92438aed9b7f4002bd9693934dbcb638 : FAILURE in 1h 49m 19s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1919c1b7bff74c438bc0861d94482329 : FAILURE in 1h 49m 53s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97aab5f8d6924b5c84271ea744211684 : FAILURE in 2h 19m 37s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c51c8f18dc8c42fa81b253e6cc5a8764 : FAILURE in 10m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d23790fc768e46b787008e4bd6d0e368 : FAILURE in 29m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6638215db706415883bdd1af0a0d516e : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4db938ed79e3ca6895a26a6e8e777d0fd0de6607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 10:26:17.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3167511118a53a07731caffd3e542e79a653d898","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 10:49:55.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9b0cfde5abb268d535cf1a76c51ccf442c7c706c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:51:37.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"cdcf53e1989fdcd358577bdbeb7651641386d979","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 10:52:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b4f7cfae0f1966c04741d67cd67bbb3510a59011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 13:17:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc7c72a983e7406fb2107531cdb22f5c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da30b65231ff4f1d80ec31f6b084bfea : SUCCESS in 12m 11s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3f76b8858a2a4f0394c1a66ce06fb226 : FAILURE in 1h 15m 13s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f8c3f0c2ce364244b6c991efd795135b : FAILURE in 2h 13m 31s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/917086e833c848499210d260c9de8632 : FAILURE in 2h 17m 01s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/df1339e02e2b4aea86ce03e42046adf2 : FAILURE in 1h 16m 46s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/725b1bdf3d964cd68ab5c9882bd225e4 : FAILURE in 2h 19m 59s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e4cb4563197841eb81b80158afc11f1d : FAILURE in 13m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/084fc7d7a9ef4dd6aca5aee7e5348f2c : FAILURE in 1h 09m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dec74b562e6441379203574c86a0197a : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":3},{"id":"cf035dfac1575f629a620f7bb8c5df7a97862334","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-20 11:35:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b3eefadd59e987da2a4d7e7ee4b09c39278970f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-20 11:49:53.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"9bcbc1e1a3f006a168f72d71e5f8d0616f12e1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 14:40:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e31e6c6c48b44838901c4bdd38fe3fd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00232af8e16a4c8cae8bf3bb8622b7ac : SUCCESS in 13m 04s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ecc5894eea8a4e04b2449e6ad52880bc : FAILURE in 1h 21m 44s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7daff068d8a84c70b6a54900fe8226cc : FAILURE in 2h 39m 08s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/78d571ce1be340fc9eebc130e3e29a8f : FAILURE in 2h 24m 24s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a4aece839a0846f5a981011f267c59fa : FAILURE in 1h 24m 08s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ac4ff0a6db14c699dc4b3bb2792ebad : FAILURE in 2h 19m 56s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9458e6304a6c477bb9849c527bd70ad2 : FAILURE in 15m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74eb1a1376664d5ca4d6de75a31c6a0b : FAILURE in 1h 09m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f06dc84d5f4555b16fca635215eeff : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":5},{"id":"49ee73aa415e5b5bf1ec08fd414696d9b5f1e023","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-01 11:08:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1325f034d0557d62bf257c5e0118f9b407b60e2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 14:17:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e744422b07a4bda8c7019920345dd6a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd3b4b0215724d1ea11b02d7a2c4cd1c : SUCCESS in 10m 22s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/77d734fa814c4ee799dd70a4d1672484 : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/47e12bae193b44c3b3a850a217c6e050 : SUCCESS in 1h 49m 07s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/64f71c054461427c9210ca631c4d97e6 : SUCCESS in 1h 27m 51s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/93731806a9b144bf95c4e5896c5ae460 : SUCCESS in 2h 17m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b4a9689702f4e8888aa4a22000fd722 : SUCCESS in 3h 05m 07s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5a3a90490b5e4d8289c44f8660e98d22 : FAILURE in 9m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d071d920cea4bcbbca3f0e6d4c7b567 : FAILURE in 36m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3daee78f78f948d3a6416114a3d11a12 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":5},{"id":"ba16ee066d2758a8be75473a55c235d72b5ee781","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-01 16:16:25.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a4b06dd62f819929269918dcd8da5e3010499201","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-05 15:28:10.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f99c1ead4a8ce16c791cc312701bde67f2dcc82f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-04-14 08:26:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"339b4be3f664f792bede9aa017c479c30dcd9069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 08:29:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3d41ef55f9a437984316ac5cbc7f01e681396535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 11:23:19.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1332acb410b84466929362220788743d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6f1405a22a64a99a07d6d8084ef1a49 : SUCCESS in 9m 02s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ee16c910628f4663a5ecf9e26551cc9a : SUCCESS in 2h 04m 49s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/421e77c5548446deb03e2d4858ce2f57 : SUCCESS in 2h 18m 41s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9ea2a254a8164f02adeed21e44e86b64 : SUCCESS in 2h 31m 29s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b7f7b689f045428088da2c039c19f8b5 : FAILURE in 1h 27m 57s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e088172d95d941f39d91464a417b0a58 : SUCCESS in 2h 52m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ccf0e2a976a48feb23df3ba751cac88 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":5},{"id":"173e785e42a4b6d6e577d00e797536d5d2739030","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-14 13:16:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0ae4057acc935f2933c3ef5fef3c09bd3723513f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 15:54:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18334b864b2046898ba7276eb6bbb3b4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3504035172714376a736ec405d84e53d : SUCCESS in 8m 10s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/68b3617483e048c9a988dae5ab4766ba : SUCCESS in 2h 12m 12s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8b937d26f72647e289cc1888a2d2191c : SUCCESS in 2h 29m 05s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a77c9a4b7d35480c8941b514f85d945e : SUCCESS in 2h 24m 55s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/213c9e80de6e42cca1dce40a3de20219 : SUCCESS in 1h 26m 21s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc9161b0b7d449278f323674da66ca8d : FAILURE in 1h 15m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1ffee0359487493caa4b4bc7c77e33e7 : FAILURE in 9m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9e4b20a1fa74403b02fddd96efdb838 : FAILURE in 1h 09m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da95d5c5665847f980cef8f549e281a3 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":5},{"id":"94dbc0f4df9aa32357bc11cc0d2425c6376bddbd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-15 20:55:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2b71d919c2ac7dbdafa9f539d82a6f9a77d96bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 23:50:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9253ebf06f489aa95fb9a5e1926880\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17d0859f952f4db09af1f18fbdd5c005 : SUCCESS in 7m 54s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bb7ec3e8494c4e12b6667a418bf2eaaf : SUCCESS in 1h 59m 08s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7cb42afd25b544afb7f1da6d88bbb0f6 : SUCCESS in 2h 01m 02s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/aa528d5fa07041608d701a92edf104cd : SUCCESS in 2h 00m 35s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/28128040f6b341b7a80a47d1416aa6dc : SUCCESS in 2h 43m 07s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fefcbf133c74f75863b7d7bcb7303e1 : SUCCESS in 1h 50m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ca76e3d668094d1b8d72b7b9fcb0c4ac : FAILURE in 10m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d40d2c371839453b9d7f25b97fdfd700 : FAILURE in 1h 04m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc0190b5c9c4d3ea45e56188df2c1ae : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":5},{"id":"40e33444a0ac6d3aed6d305bcdf3303e0466e5f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 23:53:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0f3145e4cfe0b19b993dc40dd69de64346682657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 02:30:45.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f9eec92ecb3426eb108ea8ba8c22846\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d60710f76246443cab38482dff0cdcd2 : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c8f1d08cf53c4f07b27b227643d92b3e : FAILURE in 2h 25m 45s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/58c0825ebb214091b4b566b7b87722be : SUCCESS in 2h 16m 34s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/eccc9492bb754cc79b1c85458e239dde : SUCCESS in 2h 05m 05s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8e8bde5c0ddf449cbe01f70533713ab1 : SUCCESS in 1h 57m 59s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c49beffba5c1488ebeb005045a6c12d0 : SUCCESS in 2h 31m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2d57df41f8449fb59249d7e389b676 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":5},{"id":"9e2fbae9f46ef81056c28de4b2f70074c4c4b78b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-16 08:49:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9a1531ad8980e3eb38bf3072a2838b2678059f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 12:00:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76b865a7d1a34f73ab4573477d224be8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf2fd3650d3d40af90ed19014adbb6cf : SUCCESS in 15m 44s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/abdb8d5aad9a440387e42a12adf26e6b : FAILURE in 1h 22m 53s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/04d958a35e744c64a42a6e2626999d35 : SUCCESS in 2h 34m 28s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1969cfd9917241f29b62766947c88e65 : SUCCESS in 2h 17m 38s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6e07cec1585e4a1e9164cb226fa3f111 : SUCCESS in 2h 42m 48s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/deebe4d700ed41978f4cfa0dd0a801e3 : TIMED_OUT in 3h 06m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4fdb2d8857914f0898f0e3fba339f8ec : FAILURE in 7m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0515b5ccb5344d20b38d41d15e4b57b0 : FAILURE in 31m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2509ab970eb84e78beae1e0b05d85d77 : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":5},{"id":"ac21c96fe3d4938a80b2092d480a04316bec6f47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-17 08:10:53.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"355f6308ac61a521d5b7c52679f730aaca17eeb0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-17 08:10:57.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e49f1d453b2621b19fad313f4ba8cf5ab44272c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 11:15:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4733375c0f7474bae17d33bf7850d6f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb68eb95b4734af389e491f5e475639f : SUCCESS in 10m 01s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/82097b22e6ef4066a05e2ead607b8f72 : SUCCESS in 2h 01m 27s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/800486a787ad4d5dbc7fe8052b675ac6 : SUCCESS in 2h 30m 05s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c9e0ff434fff46f1ad217394548caedd : SUCCESS in 2h 28m 11s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b0c33632fdb1406d9844bc028476d8a6 : SUCCESS in 2h 43m 50s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/569cb8b36c9048ca97b18050506478f9 : SUCCESS in 2h 56m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4168c7a83c4146319b1561fc6eaefbd9 : FAILURE in 9m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1443345de955485ca79affd64ccb2b79 : FAILURE in 1h 08m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/892c328d034a4fe49e451024b6414800 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":6},{"id":"6830653218b840c09b5aa0c4556d752284163818","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 11:18:39.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"34e9fa349879144271a2862a8f38516f707b8d89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 14:04:23.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecba2eaa59db4f7db7fcd5a79c8b7296\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96e1681d56194970a4ba894532e97e84 : SUCCESS in 9m 07s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3ca9cda04c0141ddb791ea5d3a8e2755 : SUCCESS in 2h 38m 09s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/012816df8c3b4beaac9eb06e6f44099d : SUCCESS in 2h 13m 39s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c8c5094d86f94756b97a001bb374452a : SUCCESS in 2h 28m 21s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/69064c04619a4a6a8e1128d98631b3f4 : SUCCESS in 2h 03m 33s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae09e2b56b564fd08143ede25f2509ed : SUCCESS in 2h 41m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99966338b1ee4bb18e4a409095979616 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":6},{"id":"4d120550e46d96c848fd7d10bf5c3a1caacd4482","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 14:04:23.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6},{"id":"046a88744509baa755f5b90bad744eb594361eac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 14:09:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcfd564ad09c46abb3a7faa3a32f30e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678e41e8e7c24ad4869d3a9a941eebcb : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"520591312d8941345f9fc571f15a6752880563c2","revisions":{"96a080c027177d6df3d001cc64a2e74f9e8c97c1":{"kind":"REWORK","_number":1,"created":"2026-02-24 16:56:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/977841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/41/977841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/1"}}},"commit":{"parents":[{"commit":"8ee576db9c28436ab645ed4db2ed073509aac619","subject":"Merge \"Remove deprecated run_tests/vagrantfile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/8ee576db9c28436ab645ed4db2ed073509aac619"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-24 16:56:26.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-24 16:56:26.000000000","tz":60},"subject":"Run lxc jobs instead of metal","message":"Run lxc jobs instead of metal\n\nmetal jobs are not catching ssh keypairs distribution issues,\nso it\u0027s better to run LXC jobs to ensure that role logic is not broken.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/977840\nChange-Id: I97fd4af8f70fa5468ea2b7e3213905d5d4b9d6a4\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/96a080c027177d6df3d001cc64a2e74f9e8c97c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/96a080c027177d6df3d001cc64a2e74f9e8c97c1"}]},"branch":"refs/heads/master"},"28b758b1f25d98aa46038555c72e441c36ac6c73":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-12 10:26:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/977841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/41/977841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/2"}}},"commit":{"parents":[{"commit":"8ee576db9c28436ab645ed4db2ed073509aac619","subject":"Merge \"Remove deprecated run_tests/vagrantfile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/8ee576db9c28436ab645ed4db2ed073509aac619"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-24 16:56:26.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 10:26:16.000000000","tz":0},"subject":"Run lxc jobs instead of metal","message":"Run lxc jobs instead of metal\n\nmetal jobs are not catching ssh keypairs distribution issues,\nso it\u0027s better to run LXC jobs to ensure that role logic is not broken.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980215\nChange-Id: I97fd4af8f70fa5468ea2b7e3213905d5d4b9d6a4\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/28b758b1f25d98aa46038555c72e441c36ac6c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/28b758b1f25d98aa46038555c72e441c36ac6c73"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3c3a3055ab781b5e6f07aa2a85054b3a8264cbb3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-12 10:49:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/977841/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/41/977841/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/3"}}},"commit":{"parents":[{"commit":"8ee576db9c28436ab645ed4db2ed073509aac619","subject":"Merge \"Remove deprecated run_tests/vagrantfile\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/8ee576db9c28436ab645ed4db2ed073509aac619"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-24 16:56:26.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 10:49:54.000000000","tz":0},"subject":"Run lxc jobs instead of metal","message":"Run lxc jobs instead of metal\n\nmetal jobs are not catching ssh keypairs distribution issues,\nso it\u0027s better to run LXC jobs to ensure that role logic is not broken.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: I97fd4af8f70fa5468ea2b7e3213905d5d4b9d6a4\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3c3a3055ab781b5e6f07aa2a85054b3a8264cbb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/3c3a3055ab781b5e6f07aa2a85054b3a8264cbb3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2d617acafce7e58dc55b9c72e35aa82df3f27602":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-20 11:35:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/977841/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/41/977841/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/4"}}},"commit":{"parents":[{"commit":"96cd60ecfe431a564da69a106123deed484003be","subject":"Filter swift\u0027s middleware list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/96cd60ecfe431a564da69a106123deed484003be"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-24 16:56:26.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-20 11:35:36.000000000","tz":0},"subject":"Run lxc jobs instead of metal","message":"Run lxc jobs instead of metal\n\nmetal jobs are not catching ssh keypairs distribution issues,\nso it\u0027s better to run LXC jobs to ensure that role logic is not broken.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: I97fd4af8f70fa5468ea2b7e3213905d5d4b9d6a4\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2d617acafce7e58dc55b9c72e35aa82df3f27602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2d617acafce7e58dc55b9c72e35aa82df3f27602"}]},"branch":"refs/heads/master","description":"Rebase"},"9c5a4a87c865814ca0cccc35779205fc9d2eaaea":{"kind":"NO_CHANGE","_number":5,"created":"2026-03-20 11:49:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/977841/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/41/977841/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/5"}}},"commit":{"parents":[{"commit":"be47bf4ac3ddbb59b71444a9d348fde3293dd383","subject":"Filter swift\u0027s middleware list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/be47bf4ac3ddbb59b71444a9d348fde3293dd383"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-24 16:56:26.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-20 11:49:53.000000000","tz":0},"subject":"Run lxc jobs instead of metal","message":"Run lxc jobs instead of metal\n\nmetal jobs are not catching ssh keypairs distribution issues,\nso it\u0027s better to run LXC jobs to ensure that role logic is not broken.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: I97fd4af8f70fa5468ea2b7e3213905d5d4b9d6a4\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/9c5a4a87c865814ca0cccc35779205fc9d2eaaea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/9c5a4a87c865814ca0cccc35779205fc9d2eaaea"}]},"branch":"refs/heads/master","description":"Rebase"},"520591312d8941345f9fc571f15a6752880563c2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-17 08:10:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/977841/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/41/977841/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/41/977841/6"}}},"commit":{"parents":[{"commit":"23ad4d884d9f0f5eeacb676e93fa38d91b0f2dde","subject":"Merge \"Fix post-install tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/23ad4d884d9f0f5eeacb676e93fa38d91b0f2dde"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-24 16:56:26.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-17 08:10:53.000000000","tz":0},"subject":"Run lxc jobs instead of metal","message":"Run lxc jobs instead of metal\n\nmetal jobs are not catching ssh keypairs distribution issues,\nso it\u0027s better to run LXC jobs to ensure that role logic is not broken.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/980246\nChange-Id: I97fd4af8f70fa5468ea2b7e3213905d5d4b9d6a4\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/520591312d8941345f9fc571f15a6752880563c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/520591312d8941345f9fc571f15a6752880563c2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"9c5a4a87c865814ca0cccc35779205fc9d2eaaea","theirs":"23ad4d884d9f0f5eeacb676e93fa38d91b0f2dde","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
