)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~984906","triplet_id":"openstack%2Fopenstack-ansible-os_swift~master~I67885d7a7f2734f69839a17767cede9243e0a31c","project":"openstack/openstack-ansible-os_swift","branch":"master","topic":"fix-swift-caps","attention_set":{},"removed_from_attention_set":{"39043":{"account":{"_account_id":39043,"name":"Hyunkwon Jeon","display_name":"Hyunkwon Jeon","email":"hkjeon@injeinc.co.kr","username":"hkjeon","status":"OpenStack engineer in Korea.\n"},"last_update":"2026-04-21 06:39:20.000000000","reason":"Change was submitted"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-20 10:35:06.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-21 06:39:20.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-21 06:39:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67885d7a7f2734f69839a17767cede9243e0a31c","subject":"Add AmbientCapabilities to swift_service_defaults","status":"MERGED","created":"2026-04-16 08:49:32.000000000","updated":"2026-04-21 06:40:13.000000000","submitted":"2026-04-21 06:39:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984906-fix-swift-caps","meta_rev_id":"f186197de0d4da517516aa9691406cdbe38064f9","_number":984906,"virtual_id_number":984906,"owner":{"_account_id":39043,"name":"Hyunkwon Jeon","display_name":"Hyunkwon Jeon","email":"hkjeon@injeinc.co.kr","username":"hkjeon","status":"OpenStack engineer in Korea.\n"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 06:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-04-20 10:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-04-16 08:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-20 10:02:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-04-20 10:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2026-04-16 08:55:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-20 10:02:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-16 08:55:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-16 08:59:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-04-16 08:59:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-04-16 08:59:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-04-16 08:59:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-04-16 11:37:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-20 23:19:24.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"}],"messages":[{"id":"f9ad28943dee6e1386c777ffe7531a2afb34a988","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39043,"name":"Hyunkwon Jeon","display_name":"Hyunkwon Jeon","email":"hkjeon@injeinc.co.kr","username":"hkjeon","status":"OpenStack engineer in Korea.\n"},"date":"2026-04-16 08:49:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfcc9fa959bc302999be22f8d65a3a3686640ff6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-16 08:55:21.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9aeb0e1240fe6af6d991b2c1f0b6ab7958fb8eb2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-16 08:59:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8f053b6b8f268fdc5c43873b54c7ad275f76a86b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-16 09:18:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d10092f9958b6e3c44e8e7cc869e65c2b1b8f64b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 11:37:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e98634f0aa945da96371618a9bd61f9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52cdac8cd84e4c418249dd408b1c5593 : SUCCESS in 12m 24s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c21a145e032a4b7686e7470bc9eedcec : SUCCESS in 53m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/25e23f1eb6f64de2839ee69faa4acd6d : SUCCESS in 1h 40m 29s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b501adf6897f4ba3a65b594cc07d7177 : SUCCESS in 1h 49m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bea2350234174af3a262ff968b40d2cd : SUCCESS in 51m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e4214aa1dd094f99a0faea3e8ae6bd7c : SUCCESS in 1h 27m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de0747c15509478aace0fa349af5c35a : SUCCESS in 59m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6fc28811896487181c548ea239d5922 : SUCCESS in 1h 46m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/70f660ecd418476ca1eac9667d5f1f77 : SUCCESS in 1h 17m 21s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ac8f67736159463c9a719166ab1caf32 : SUCCESS in 2h 43m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2746b26f63f4efcbfe44822a9efc39f : FAILURE in 28m 28s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37b1f4f0b27e46a993815375fbce2728 : SUCCESS in 2h 12m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3086a15e605b4429a24e9f56787c1445 : SUCCESS in 2h 34m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/915980f1eb1b49f681c8ac2b95f49bb5 : FAILURE in 15m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efd8b4aca49e4183a8cecbadff08b25e : FAILURE in 1h 09m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e264d07ef764890872cb3978838306d : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2091c457321a516d56ec7d3690e274fcc9b7590b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-16 12:52:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f9e4c6541ee7e8be60731a462f3d9e15ef042a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 15:19:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d48fcefed1c40459717764a90a034b7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a8a6561e4794921a0a9e1f3fcb9bfab : SUCCESS in 9m 04s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/27dbfa962b4545dbb68b54b957256217 : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6696aa8cb2bd42019000f674ed714a65 : SUCCESS in 1h 32m 10s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a3d3417f65fa475b9ff60342cc94e314 : SUCCESS in 54m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8db08cf01d5b478da605c2475a310a12 : SUCCESS in 51m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0b02734473e74ae085ade23cb8200aa2 : SUCCESS in 1h 18m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfa9911fd6ca4f8fba616c62ad0e6eb6 : SUCCESS in 1h 34m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83f95ce7e3a34908b85870bee0f555a0 : SUCCESS in 1h 16m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5f3ba3a384bc4bd6937580649946aaac : SUCCESS in 1h 53m 05s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8b732bb3839f4ce9ae79fea50666ebf7 : SUCCESS in 1h 55m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab5d1c3d3d0e4c829d41fe7ab0a4fcb3 : SUCCESS in 2h 15m 36s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c51ef8ab89f9427e88e90a1c46fae65b : SUCCESS in 2h 20m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c5983caad714825abf64071aa2aa3fe : SUCCESS in 2h 04m 58s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cc9f31e15ac24b9597c883229465bd33 : FAILURE in 7m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8855047f4464f6ca23a0c409c5445aa : FAILURE in 50m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb7f5522f474c16918a07f5c285b1fe : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f1cecdedcab4f9aa905378849ff22bc8724c9f75","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-20 10:02:34.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b5a84a46082936b980037d7bd474453c5ff56bc","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-04-20 10:35:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5287c133498e2f17d5ccb2ed0b650b9610e41851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:38:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee4eca552d41ad703d030f7d7c5654bf997ce2a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 13:34:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81435dca3cae468f814f8a2c8d4c466b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad0613442a9244f88286c7d51509e029 : SUCCESS in 15m 07s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9e58d8a06fb74a60a3d8c3cc352931a7 : FAILURE in 1h 24m 02s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/03ff0dc363ac4473aece35e1c019408a : SUCCESS in 2h 18m 20s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/172a99db71bd401f9cd7a952fef90dc3 : SUCCESS in 1h 28m 31s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/568841f719fb48149a4858ee6467d588 : SUCCESS in 2h 08m 30s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5e77be4bd7d49c7846086f55fc34267 : SUCCESS in 2h 49m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d9a4f5d8dd8438894d9b8e3a0dd1749 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4b17d82fdf70e5f0c90d9ee21290a4f32f5f54a8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-20 14:59:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccd94fd5faade15246c5cc86c0c413ba20830a47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 17:56:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7a6153a57694a6f8dde1c0610436973\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/784bed31a07040ecb931613561b574fb : SUCCESS in 13m 23s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d2d4ecdcf85449c982a120dba0f7392c : SUCCESS in 2h 09m 11s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ce013d07d615482089941dd72a051ff6 : SUCCESS in 2h 27m 40s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ea34487099a548d9a58fb353ea3bc7d8 : SUCCESS in 1h 37m 41s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/36ea87fab74940aa9945f641a86225c6 : SUCCESS in 1h 57m 03s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe9ca5051a2641e294d9b613991b8ba7 : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/79221a894e96479c8767fed71f0aa056 : FAILURE in 19m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74b3be2662bd4a73b63a98bc197c1b06 : FAILURE in 1h 01m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0547aaa70e754589887aad4e034a89a5 : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4139d88f8d3711a028455d54afef5e40dfcdf0b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 17:58:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c1657c45d7447ab7a2bb507ef62d4f46b4209ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 20:42:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e02df87b1d17458687e8211643749b96\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e36489f3d8f546c99f2cde5ad1441f62 : SUCCESS in 13m 23s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/08160e57886f4951b9c33688e1161618 : SUCCESS in 2h 33m 03s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c2dea230826343178608a4a3ff70e6c0 : SUCCESS in 2h 18m 01s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0b392731b3ae4e3ab4644b53314d969d : SUCCESS in 2h 35m 22s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f3836aacefe84501a3cb91f6486cd40a : FAILURE in 1h 38m 22s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44690a9fe57844539192e0c207d529ad : SUCCESS in 2h 38m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0047b1c63a41426fbd66badd01b3b346 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"44d5c68931fd3811bb92c79fa88ea3c5122ebd2b","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-04-20 23:19:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45e5500fff99a15aa23d4e343b83c8b003472cae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 03:52:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d285b964e354b6db2ecb5837c904653\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbdaa84f0af94ba7aa6ee2abc9012b3e : SUCCESS in 17m 14s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/29dda92c42554d4da3cf274658982ca8 : SUCCESS in 2h 21m 56s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8fe8b548b08e45a7b8d231950a5e73a6 : SUCCESS in 2h 21m 20s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e6f590f0ea4d4c7097448bb2631a6a0a : SUCCESS in 2h 31m 59s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3116af9435304bff8ce0750f6c4d8d14 : SUCCESS in 2h 37m 38s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aba77272332491a890d46d569f46a0c : SUCCESS in 3h 00m 41s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/31229a51e83b4cf38afe46739c8f52ec : FAILURE in 12m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/421171f1e33348e38d86ed40be80d0bc : FAILURE in 1h 02m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ca7f1bb36a44d481a9c90210ff852a : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6dc2306f3966f96715c2ede980a87e798210f603","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 03:55:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4eecdbab7c8d1632303aee7540b7e91300d5b211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 06:39:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec6fca2738dc478cb30be639343bc5e8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8833dd03dec749b39bd0dd39b9779566 : SUCCESS in 7m 41s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/555881ce7fc24dc1a6c0f626266f56d9 : SUCCESS in 1h 16m 59s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d9c8e04a93ff4826a4a21eac22ee210a : SUCCESS in 1h 15m 57s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/184ccea75e7948dd978540d72d6ac2a8 : SUCCESS in 2h 16m 38s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5d83baedb8c1450387a4e148dc4f0eb4 : SUCCESS in 2h 42m 55s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b947157df9e48d39deb6303cadefb7d : SUCCESS in 1h 15m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7215890abdf04c9db7104d0f4b5a68b7 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"052821781f01f2d12a063150a85d7bc4d28320e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 06:39:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f186197de0d4da517516aa9691406cdbe38064f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 06:40:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a99ff2af60e4cba97d36027d604f6b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ee07cf70d54cfcbd22eb2e6f5d5802 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21daafb630f9d9202f977e72d8515a0503d45e09","revisions":{"21daafb630f9d9202f977e72d8515a0503d45e09":{"kind":"REWORK","_number":1,"created":"2026-04-16 08:49:32.000000000","uploader":{"_account_id":39043,"name":"Hyunkwon Jeon","display_name":"Hyunkwon Jeon","email":"hkjeon@injeinc.co.kr","username":"hkjeon","status":"OpenStack engineer in Korea.\n"},"ref":"refs/changes/06/984906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/06/984906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/06/984906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/06/984906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/06/984906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/06/984906/1"}}},"commit":{"parents":[{"commit":"23ad4d884d9f0f5eeacb676e93fa38d91b0f2dde","subject":"Merge \"Fix post-install tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/23ad4d884d9f0f5eeacb676e93fa38d91b0f2dde"}]}],"author":{"name":"Hyunkwon Jeon","email":"nosmile0412@hanmail.net","date":"2026-04-16 08:49:04.000000000","tz":540},"committer":{"name":"Hyunkwon Jeon","email":"nosmile0412@hanmail.net","date":"2026-04-16 08:49:04.000000000","tz":540},"subject":"Add AmbientCapabilities to swift_service_defaults","message":"Add AmbientCapabilities to swift_service_defaults\n\nThis commit adds `AmbientCapabilities: \"CAP_SETGID CAP_SETUID\"`\nto the `swift_service_defaults` configuration.\n\nWithout these capabilities, the Swift daemon fails to start\ndue to a `PermissionError on os.setgid()`. Adding them ensures\nthe daemon runs correctly with the required permissions.\n\nChange-Id: I67885d7a7f2734f69839a17767cede9243e0a31c\nSigned-off-by: Hyunkwon Jeon \u003cnosmile0412@hanmail.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/21daafb630f9d9202f977e72d8515a0503d45e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/21daafb630f9d9202f977e72d8515a0503d45e09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
