)]}'
{"id":"openstack%2Fopenstack-ansible-os_swift~985624","triplet_id":"openstack%2Fopenstack-ansible-os_swift~stable%2F2025.1~I67885d7a7f2734f69839a17767cede9243e0a31c","project":"openstack/openstack-ansible-os_swift","branch":"stable/2025.1","topic":"fix-swift-caps-stable/2025.2-stable/2025.1","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-22 09:50:40.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-22 12:57:13.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-22 12:57:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67885d7a7f2734f69839a17767cede9243e0a31c","subject":"Add AmbientCapabilities to swift_service_defaults","status":"MERGED","created":"2026-04-21 07:31:23.000000000","updated":"2026-04-22 12:58:08.000000000","submitted":"2026-04-22 12:57:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985624-fix-swift-caps-stable/2025.2-stable/2025.1","cherry_pick_of_change":985623,"cherry_pick_of_patch_set":1,"meta_rev_id":"570de06b79ad9f2d1f98b0e12da8d1077a191806","_number":985624,"virtual_id_number":985624,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-22 12:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-04-22 09:50:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-04-21 09:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-04-22 09:50:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":39043,"name":"Hyunkwon Jeon","display_name":"Hyunkwon Jeon","email":"hkjeon@injeinc.co.kr","username":"hkjeon","status":"OpenStack engineer in Korea.\n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-21 07:31:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":39043,"name":"Hyunkwon Jeon","display_name":"Hyunkwon Jeon","email":"hkjeon@injeinc.co.kr","username":"hkjeon","status":"OpenStack engineer in Korea.\n"},"state":"CC"},{"updated":"2026-04-21 09:46:54.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-21 10:32:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 10:36:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-04-21 10:36:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"699be98fb4e215eb8e2aa4600b57196b4c39819c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-21 07:31:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"6d25d0ec2b11295ddd7c3715128785d2de8f7ef7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-21 09:46:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f56f2ea9f2facfec9c8f007e98831098294249db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 10:32:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50bc4dade0cd4e22bc0b4df38868a7cd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/095b85e81f2b4758b2498c5537d8b1ae : SUCCESS in 16m 57s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/42288553792643eb8a5d0192470e9dea : SUCCESS in 1h 35m 50s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ced7bd1d9ccf4090a0513830cfcadbee : SUCCESS in 1h 32m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/37364e4eaf1d4dd48dbff8bbeffb67f0 : SUCCESS in 1h 33m 21s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a5b72373bab3406ab3d7228eca5e30c7 : SUCCESS in 1h 38m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9d61250b0fb246beabe11ec4210cfcbf : SUCCESS in 1h 44m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b944cad16c4346ff836c21f267fecd3f : SUCCESS in 48m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c1db9eb232445b3a4ca70041b14abd6 : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efc8ed24c49f4341b9038eff9f57d23c : SUCCESS in 1h 34m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/58503a665e894313a85b702eaa4f9e4a : SUCCESS in 1h 40m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3dfd66f971d7489684a4d349776da271 : SUCCESS in 1h 22m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5a1aeb3f2b447ce9ad0ba96800d105e : SUCCESS in 1h 30m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c15d1501f0a47bdb886485564b51436 : SUCCESS in 2h 22m 10s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2c08ddf149524218b37bd52716aef3c6 : SUCCESS in 2h 32m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71e163723ab145939be321c3c2233e2b : SUCCESS in 2h 18m 19s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aacda1c9cdf44fb7bc2b3f82b0efab9a : SUCCESS in 58m 33s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f690571788654503a10df1318e9684fb : SUCCESS in 2h 34m 48s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68aeefce66964b65bab2a7e47cfb5077 : SUCCESS in 2h 25m 44s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/99f4aab76b304dc19aa38405fea6b72d : SUCCESS in 1h 16m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b4750a2f2a1845328d4863688169c721 : FAILURE in 8m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7dd7401416943ee8cdca9952fb10470 : SUCCESS in 1h 14m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544a928e24c44ca1bb1059e3f6c48a0b : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ef36c541b0d711bdee7b06de8daddd84abfab667","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-04-22 09:50:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04c88a0c021975827bd1723b8cf72e34ca5b596d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 09:54:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"919d0454f56d6be533e718bd0529573a8fbf4974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 12:57:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5db7139ccc4902882d9d6aecb9a876\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5db0aff99e7d4b52b7bf02bf8cc099f4 : SUCCESS in 13m 18s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d30c2a67bd6b4efd9b99e3ca0bffb4cd : SUCCESS in 1h 38m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/552e6fb97cf54ef9b99b951347cd968b : SUCCESS in 1h 31m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7e55f5ff59a493cb93dd0e8247fede6 : SUCCESS in 1h 23m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/85a68df8bdf3429c8f33e7fffc079c51 : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9fb0cf2d2c92477d9272469cf1016e28 : SUCCESS in 1h 23m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1a0fab6039a49a79fa267e90ce4c727 : SUCCESS in 1h 38m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8be1b5eccdc8475f86107fca303bc1c5 : SUCCESS in 1h 30m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f0a9aea4f164058873a75930920620b : SUCCESS in 1h 40m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c8dfaec0e2a43d4974c44b3f5ace422 : SUCCESS in 1h 25m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d51a564e6fb64bc59abd97ed55bc80b9 : SUCCESS in 1h 41m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97a7058b225e417492a9f1428e777646 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0cf4599171d18cc7b7153795c7afb9193bd434e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 12:57:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"570de06b79ad9f2d1f98b0e12da8d1077a191806","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 12:58:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4dd3dad080f4d98a5e596636a1540ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2590e828e1f94456be7a1a46bae20064 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"542f77a111d91a608f2781047212b82a0c41438d","revisions":{"542f77a111d91a608f2781047212b82a0c41438d":{"kind":"REWORK","_number":1,"created":"2026-04-21 07:31:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/985624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_swift","ref":"refs/changes/24/985624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/985624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/985624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/985624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_swift refs/changes/24/985624/1"}}},"commit":{"parents":[{"commit":"2f1b8aeda606bec0eb7b08dca08c9bbdcfefc63e","subject":"Fix swift service default resolution","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/2f1b8aeda606bec0eb7b08dca08c9bbdcfefc63e"}]}],"author":{"name":"Hyunkwon Jeon","email":"nosmile0412@hanmail.net","date":"2026-04-16 08:49:04.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-21 07:31:23.000000000","tz":0},"subject":"Add AmbientCapabilities to swift_service_defaults","message":"Add AmbientCapabilities to swift_service_defaults\n\nThis commit adds `AmbientCapabilities: \"CAP_SETGID CAP_SETUID\"`\nto the `swift_service_defaults` configuration.\n\nWithout these capabilities, the Swift daemon fails to start\ndue to a `PermissionError on os.setgid()`. Adding them ensures\nthe daemon runs correctly with the required permissions.\n\nChange-Id: I67885d7a7f2734f69839a17767cede9243e0a31c\nSigned-off-by: Hyunkwon Jeon \u003cnosmile0412@hanmail.net\u003e\n(cherry picked from commit 21daafb630f9d9202f977e72d8515a0503d45e09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/542f77a111d91a608f2781047212b82a0c41438d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_swift/commit/542f77a111d91a608f2781047212b82a0c41438d"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"2f1b8aeda606bec0eb7b08dca08c9bbdcfefc63e","theirs":"b62cb5bb2e2af0cd46e39b33afb0cbdd46328731","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
