)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~650054","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"improv","hashtags":[],"change_id":"I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5","subject":"Switch to import_task in os_tempest","status":"MERGED","created":"2019-04-04 15:39:46.000000000","updated":"2019-06-12 14:55:17.000000000","submitted":"2019-06-12 14:55:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650054-1560351317869-22ee9161","meta_rev_id":"7d4d60612e00834105fa60ce30013ef73efcdac1","_number":650054,"virtual_id_number":650054,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"date":"2019-06-12 11:49:52.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-12 14:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-06-12 09:02:03.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":1,"date":"2019-06-12 09:03:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-06-12 09:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-06-12 09:51:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":27379,"name":"jrosser","inactive":true},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-04 15:39:46.000000000","updated_by":{"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},"reviewer":{"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},"state":"REVIEWER"},{"updated":"2019-04-04 15:40:13.000000000","updated_by":{"_account_id":27379,"name":"jrosser","inactive":true},"reviewer":{"_account_id":27379,"name":"jrosser","inactive":true},"state":"REVIEWER"},{"updated":"2019-04-04 15:40:28.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-10 21:19:42.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 12:31:23.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-06-11 12:39:35.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2019-06-11 15:09:21.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-06-12 09:02:03.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-06-12 09:03:00.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-06-12 09:51:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-12 14:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2cd6a234247f61b5979ce689fbe0a27d624a184d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-04 15:39:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9603fc884e69e00c02c2dc8718464b2de339fa7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 17:04:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/1/check/openstack-ansible-linters/9a1c25d/ : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/1/check/openstack-ansible-deploy-aio_metal-centos-7/92d0ad1/ : FAILURE in 1h 01m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c75e9a9/ : FAILURE in 47m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/04a739b/ : FAILURE in 45m 30s\n- openstack-tox-docs http://logs.openstack.org/54/650054/1/check/openstack-tox-docs/2fcd0ac/html/ : SUCCESS in 6m 03s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/54/650054/1/check/openstack-ansible-functional-tempestconf-centos-7/feeaaf3/ : FAILURE in 1h 20m 15s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/1/check/tripleo-ci-centos-7-standalone-os-tempest/b8b2115/ : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":1},{"id":"570ec36d37acc1465b9660840840df2ec5f70ce9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-05 11:12:54.000000000","message":"Abandoned\n\nNot needed currently.","accounts_in_message":[],"_revision_number":1},{"id":"c46f6caae4b55fd4bbecad6b6da01adc7c7ef99c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-05 11:13:03.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"74ffa41ad0d36956e83925fb2c92d7cd3fe7b153","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 12:26:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/1/check/openstack-ansible-linters/1929f0b/ : SUCCESS in 5m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/1/check/openstack-ansible-deploy-aio_metal-centos-7/2bec323/ : FAILURE in 1h 02m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b1c9eb2/ : FAILURE in 49m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/d254b90/ : FAILURE in 43m 41s\n- openstack-tox-docs http://logs.openstack.org/54/650054/1/check/openstack-tox-docs/ad24857/html/ : SUCCESS in 2m 46s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/54/650054/1/check/openstack-ansible-functional-tempestconf-centos-7/2cba637/ : FAILURE in 1h 08m 08s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/1/check/tripleo-ci-centos-7-standalone-os-tempest/c76e4f1/ : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":1},{"id":"17987ea391307afd1b79da951065da4cdc95fe56","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-05 12:28:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b2e3ba2b407e094ec32c2d374e170f40fc282b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 13:46:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/2/check/openstack-ansible-linters/928980d/ : SUCCESS in 5m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/2/check/openstack-ansible-deploy-aio_metal-centos-7/4129f5f/ : FAILURE in 1h 05m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0590a0f/ : FAILURE in 47m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/75745da/ : FAILURE in 49m 41s\n- openstack-tox-docs http://logs.openstack.org/54/650054/2/check/openstack-tox-docs/bbd4f2f/html/ : SUCCESS in 3m 00s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/54/650054/2/check/openstack-ansible-functional-tempestconf-centos-7/6129632/ : FAILURE in 1h 13m 01s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/2/check/tripleo-ci-centos-7-standalone-os-tempest/ce81b9e/ : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e76df1355f5ffe1936a3760afcbf9dc0f1b1f638","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-04-05 14:07:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cab8c51a4e6b9eadf53858b7e000c00fb33f7c71","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-04-06 23:54:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nAll failure jobs have: The task includes an option with an undefined variable","accounts_in_message":[],"_revision_number":2},{"id":"277ac22d4676bf6cec297da67cb4b8d2679086d1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-08 03:28:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"38f281efabc1baa0a88b5b0a57abe66b4f97f618","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 04:41:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/3/check/openstack-ansible-linters/307995c/ : SUCCESS in 4m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/3/check/openstack-ansible-deploy-aio_metal-centos-7/6b2c247/ : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2fb96a4/ : SUCCESS in 58m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/a40402a/ : SUCCESS in 52m 04s\n- openstack-tox-docs http://logs.openstack.org/54/650054/3/check/openstack-tox-docs/c16f586/html/ : SUCCESS in 2m 48s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/54/650054/3/check/openstack-ansible-functional-tempestconf-centos-7/83c496b/ : FAILURE in 58m 25s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/3/check/tripleo-ci-centos-7-standalone-os-tempest/05002ca/ : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":3},{"id":"278161b38cce6dae7ffd52b8ff0c3993379b14ab","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-08 09:28:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4eebce2f316305422b291523f3282fd0a6a55d26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 11:15:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/4/check/openstack-ansible-linters/e1d40a3/ : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/4/check/openstack-ansible-deploy-aio_metal-centos-7/5b1b39c/ : SUCCESS in 1h 16m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/4d76b5b/ : SUCCESS in 57m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/c784a82/ : SUCCESS in 52m 49s\n- openstack-tox-docs http://logs.openstack.org/54/650054/4/check/openstack-tox-docs/1c4ec8b/html/ : SUCCESS in 3m 04s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/54/650054/4/check/openstack-ansible-functional-tempestconf-centos-7/a34a977/ : FAILURE in 1h 12m 29s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/4/check/tripleo-ci-centos-7-standalone-os-tempest/6ee5c1a/ : SUCCESS in 1h 12m 10s","accounts_in_message":[],"_revision_number":4},{"id":"67288ed9ade1dbec05a21d8493e88990f67d43bb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-10 13:43:27.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"c1d13401429a276d771ad08c823a56ce32dbdc18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 15:09:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/5/check/openstack-ansible-linters/bbae492/ : SUCCESS in 6m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/5/check/openstack-ansible-deploy-aio_metal-debian-stable/f6e4cf3/ : SUCCESS in 58m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/5/check/openstack-ansible-deploy-aio_metal-centos-7/ecd3fda/ : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/5/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/17e3884/ : SUCCESS in 57m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/5/check/openstack-ansible-deploy-aio_distro_metal-centos-7/ebbb425/ : SUCCESS in 55m 47s\n- openstack-tox-docs http://logs.openstack.org/54/650054/5/check/openstack-tox-docs/a5de940/html/ : SUCCESS in 3m 08s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/54/650054/5/check/openstack-ansible-functional-tempestconf-centos-7/ec54d1a/ : FAILURE in 1h 18m 44s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/5/check/tripleo-ci-centos-7-standalone-os-tempest/483e172/ : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":5},{"id":"bfbdde49052e4f8ba5d974b51efcb83c77fcaf4d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-11 08:01:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"76f69043299e9dcc385c57fd0d3340932302532b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-11 08:02:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"348e72320c07618de1d42e5a76fea479a2bd3e44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 09:23:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/7/check/openstack-ansible-linters/4054a23/ : SUCCESS in 8m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/7/check/openstack-ansible-deploy-aio_metal-debian-stable/fca1b99/ : SUCCESS in 59m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/7/check/openstack-ansible-deploy-aio_metal-centos-7/9a4b88c/ : SUCCESS in 1h 14m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/7/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e704875/ : SUCCESS in 1h 05m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/7/check/openstack-ansible-deploy-aio_distro_metal-centos-7/bc05d51/ : SUCCESS in 1h 00m 39s\n- openstack-tox-docs http://logs.openstack.org/54/650054/7/check/openstack-tox-docs/bd52b16/html/ : SUCCESS in 4m 47s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/54/650054/7/check/openstack-ansible-functional-tempestconf-centos-7/d0f58e9/ : FAILURE in 1h 15m 17s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/7/check/tripleo-ci-centos-7-standalone-os-tempest/1ed74b6/ : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":7},{"id":"71e9c735afd020fb314af5f8c54532d041502582","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-12 09:07:24.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0a245300c638b3a2496e69833ad937ece0cc9bae","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-12 09:27:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"41930595deaacb0c28885d847a3c39eea36857bf","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-12 09:27:43.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ce4fa42cee4e4dd782b971d57993bef1f335afc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 10:44:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/8/check/openstack-ansible-linters/30daf47/ : SUCCESS in 5m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/8/check/openstack-ansible-deploy-aio_metal-debian-stable/b95a574/ : SUCCESS in 1h 00m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/8/check/openstack-ansible-deploy-aio_metal-centos-7/74cd9d1/ : SUCCESS in 1h 13m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/8/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5e0012a/ : SUCCESS in 55m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/8/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8785111/ : SUCCESS in 45m 22s\n- openstack-tox-docs http://logs.openstack.org/54/650054/8/check/openstack-tox-docs/fa3764b/html/ : SUCCESS in 3m 09s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/54/650054/8/check/openstack-ansible-functional-tempestconf-centos-7/5cc3faa/ : FAILURE in 1h 06m 08s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/8/check/tripleo-ci-centos-7-standalone-os-tempest/19985e9/ : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":8},{"id":"99ff9447ed3758075b0feb538c8ff212274997f4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-15 12:56:43.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"95926aba1aa0211f3f72862adc9368f9a292000b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 14:20:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/9/check/openstack-ansible-linters/d06daa9/ : SUCCESS in 6m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/9/check/openstack-ansible-deploy-aio_metal-debian-stable/85062cb/ : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/9/check/openstack-ansible-deploy-aio_metal-centos-7/794e3b8/ : SUCCESS in 1h 16m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/9/check/openstack-ansible-deploy-aio_metal-opensuse-150/fbb36ae/ : SUCCESS in 1h 04m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/cb0ee3f/ : SUCCESS in 57m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/decc873/ : SUCCESS in 57m 11s\n- openstack-tox-docs http://logs.openstack.org/54/650054/9/check/openstack-tox-docs/910e9ff/html/ : SUCCESS in 3m 29s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/54/650054/9/check/openstack-ansible-functional-tempestconf-centos-7/e241660/ : FAILURE in 1h 16m 18s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/9/check/tripleo-ci-centos-7-standalone-os-tempest/0c2207d/ : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":9},{"id":"14525d376f344376b175d4cdf314b52a5583ce65","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-04 11:58:18.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"79f4af805af87b56eae072a30ad9af6291db8c3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 13:29:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/10/check/openstack-ansible-linters/9fbc881/ : SUCCESS in 7m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_metal-debian-stable/1b5e825/ : SUCCESS in 1h 04m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_metal-centos-7/26a6677/ : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_metal-opensuse-150/e89550f/ : SUCCESS in 1h 05m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/23239f2/ : RETRY_LIMIT in 1m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_distro_metal-centos-7/6ae56a1/ : SUCCESS in 58m 56s\n- openstack-tox-docs http://logs.openstack.org/54/650054/10/check/openstack-tox-docs/24ccab3/html/ : SUCCESS in 2m 38s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/1e03288/ : SUCCESS in 1h 06m 39s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/10/check/tripleo-ci-centos-7-standalone-os-tempest/de09168/ : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":10},{"id":"3648073ff65e569e5a3aa27860e8a455affa4088","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-10 07:28:50.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"10a7f6f02dde4aa735cc8a6475c39c9b3908c75c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 09:15:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/10/check/openstack-ansible-linters/cf7ea71/ : SUCCESS in 5m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_metal-debian-stable/1c55659/ : SUCCESS in 1h 26m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_metal-centos-7/3fb3877/ : SUCCESS in 1h 40m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_metal-opensuse-150/6f884f0/ : SUCCESS in 1h 19m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b03de3f/ : SUCCESS in 1h 01m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_distro_metal-centos-7/1b2c4f1/ : SUCCESS in 1h 03m 34s\n- openstack-tox-docs http://logs.openstack.org/54/650054/10/check/openstack-tox-docs/a982f6e/html/ : SUCCESS in 2m 49s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/54/650054/10/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/85a4419/ : SUCCESS in 1h 02m 50s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/10/check/tripleo-ci-centos-7-standalone-os-tempest/72391b1/ : SUCCESS in 1h 12m 46s","accounts_in_message":[],"_revision_number":10},{"id":"0bc70157960b940d866dc3db0a164ab2ec2f45e1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 10:00:20.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a40ae42edb71c031436112cb3009e0b6b51e88cd","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-10 10:06:01.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"747ded28a287229df8efbd1e4bf2837f494a6006","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-10 10:06:06.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"088b0d7272bd8930deeff02fb69946642437df05","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 10:09:18.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"6a8405f37e843fb1abcbc614dd0e21e7f318a55d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 11:51:39.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/11/check/openstack-ansible-linters/29dbe76/ : SUCCESS in 5m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/11/check/openstack-ansible-deploy-aio_metal-debian-stable/07a2de9/ : SUCCESS in 1h 05m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/11/check/openstack-ansible-deploy-aio_metal-centos-7/fae2937/ : SUCCESS in 1h 43m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/11/check/openstack-ansible-deploy-aio_metal-opensuse-150/eb7d83a/ : SUCCESS in 1h 09m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/11/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dd9d277/ : SUCCESS in 1h 00m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/11/check/openstack-ansible-deploy-aio_distro_metal-centos-7/aa08ca5/ : SUCCESS in 1h 02m 01s\n- openstack-tox-docs http://logs.openstack.org/54/650054/11/check/openstack-tox-docs/9e880bc/html/ : SUCCESS in 2m 40s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/54/650054/11/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/8d0f398/ : SUCCESS in 1h 00m 09s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/11/check/tripleo-ci-centos-7-standalone-os-tempest/17a3b2c/ : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":11},{"id":"d929cf35d679b589867d8e60605da187f1cdb214","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-10 21:19:42.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7d8c72e3b809acdb13b22f31b9943e1942917c78","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-11 06:00:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"67fecba25a247aa3ce06a8dd66295daeac0c2bec","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-11 06:00:05.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"72915c10d4497b5f2aa7e8de620ff72128a5aa8c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-11 06:19:03.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"b656d6533df289300fe5fe2ce0b9ec2254dd56d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 08:10:34.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/12/check/openstack-ansible-linters/0cd7978/ : SUCCESS in 4m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/12/check/openstack-ansible-deploy-aio_metal-debian-stable/eb6f114/ : SUCCESS in 1h 01m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/12/check/openstack-ansible-deploy-aio_metal-centos-7/7ca818b/ : SUCCESS in 1h 21m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/12/check/openstack-ansible-deploy-aio_metal-opensuse-150/96d561d/ : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/12/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/7a460dd/ : SUCCESS in 56m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/12/check/openstack-ansible-deploy-aio_distro_metal-centos-7/f4275c7/ : SUCCESS in 1h 00m 56s\n- openstack-tox-docs http://logs.openstack.org/54/650054/12/check/openstack-tox-docs/655fe52/html/ : SUCCESS in 2m 54s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/54/650054/12/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/be4217a/ : SUCCESS in 1h 09m 25s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/12/check/tripleo-ci-centos-7-standalone-os-tempest/58d2b91/ : SUCCESS in 1h 28m 05s","accounts_in_message":[],"_revision_number":12},{"id":"304ef8133acb26a412200003f3c8a1eead4a40fd","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-06-11 08:31:05.000000000","message":"Patch Set 12:\n\nThe commit message says that include_role is replaced by import_role .. however, I found include_role in tasks/tempest_install.yml and tasks/tempestconf.yml","accounts_in_message":[],"_revision_number":12},{"id":"fd37ab3ec8da7ccb272541f82bcf8150025a1cba","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-11 10:23:12.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b2cdfb799a362710f1d2ee725b006b47ebc333a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 11:58:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/13/check/openstack-ansible-linters/81c4470/ : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/13/check/openstack-ansible-deploy-aio_metal-debian-stable/7465ded/ : SUCCESS in 1h 03m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/13/check/openstack-ansible-deploy-aio_metal-centos-7/80b7db1/ : SUCCESS in 1h 16m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/13/check/openstack-ansible-deploy-aio_metal-opensuse-150/046e319/ : SUCCESS in 1h 16m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/13/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/433e275/ : SUCCESS in 58m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/13/check/openstack-ansible-deploy-aio_distro_metal-centos-7/b750453/ : SUCCESS in 1h 00m 10s\n- openstack-tox-docs http://logs.openstack.org/54/650054/13/check/openstack-tox-docs/25b6dc7/html/ : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/54/650054/13/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/717894d/ : SUCCESS in 1h 00m 12s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/13/check/tripleo-ci-centos-7-standalone-os-tempest/20c61a3/ : SUCCESS in 1h 27m 06s","accounts_in_message":[],"_revision_number":13},{"id":"c7b2aa9cccefa9dbab8c8982d4c03f0bc0f07918","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-06-11 12:31:23.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"b64d030506843c8d1d466c5fdebeeebe84949083","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-06-11 12:39:35.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"7dbb95c8121b38717648155417d9911184313cdd","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-11 15:09:21.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)\n\none question, but otherwise looks good to me","accounts_in_message":[],"_revision_number":13},{"id":"76cbe6ef78676bebcd31d4a7190c761b488435aa","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-11 18:19:21.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"be4e26e6a8a711615fb896d9b67576f9a660a0d5","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-11 20:01:16.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"7f4459482719964ee407c2f2d47bb01917702dd1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-12 06:33:19.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"7c60447356b5a486468a9d131776a603a8665a3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 08:53:48.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/14/check/openstack-ansible-linters/ecdae4f/ : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_metal-debian-stable/a1af28b/ : SUCCESS in 59m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_metal-centos-7/81f677a/ : SUCCESS in 1h 38m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_metal-opensuse-150/368a56f/ : SUCCESS in 1h 02m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b827239/ : SUCCESS in 57m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_distro_metal-centos-7/f22aed2/ : SUCCESS in 56m 06s\n- openstack-tox-docs http://logs.openstack.org/54/650054/14/check/openstack-tox-docs/a5fe0ae/html/ : SUCCESS in 4m 08s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/d5102b6/ : SUCCESS in 1h 07m 22s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/14/check/tripleo-ci-centos-7-standalone-os-tempest/cbe9f83/ : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":14},{"id":"cb58f123a2ffcca1a2b3a7ccea7a2544c2b3e11e","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-06-12 09:02:03.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"2f923249ea79c02016c8d8ce3065b9a6bfc4cd51","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-06-12 09:03:00.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"4fb889517224035ceec6c498ab901e3210d380f6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-12 09:51:28.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"3d8e5a69a784bb26ab0bf04ec6b85c066f646ea6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 09:51:40.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"19c71fa0e3a6d6deb4b72cc5e4ad7fe64ff79666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 11:26:06.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/14/gate/openstack-ansible-linters/d42662c/ : SUCCESS in 4m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_metal-debian-stable/100150b/ : SUCCESS in 59m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_metal-centos-7/5537219/ : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_metal-opensuse-150/3e7ac7e/ : FAILURE in 42m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5085f23/ : SUCCESS in 59m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/f564c5e/ : SUCCESS in 1h 06m 58s\n- openstack-tox-docs http://logs.openstack.org/54/650054/14/gate/openstack-tox-docs/5852f97/html/ : SUCCESS in 2m 53s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/696c39d/ : SUCCESS in 1h 02m 08s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/14/gate/tripleo-ci-centos-7-standalone-os-tempest/3723579/ : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":14},{"id":"195dd298379ad309d82d83525dd56178d2c50606","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-12 11:49:52.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"8b6d0b74d0596c23b0d7ea8c7922a5bdb8650ac4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 13:19:28.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/14/check/openstack-ansible-linters/ffc1175/ : SUCCESS in 8m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_metal-debian-stable/2eb4302/ : SUCCESS in 1h 00m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_metal-centos-7/cd338e5/ : SUCCESS in 1h 18m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_metal-opensuse-150/a1bd3c6/ : SUCCESS in 1h 05m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/166ad67/ : SUCCESS in 57m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_distro_metal-centos-7/70fce91/ : SUCCESS in 56m 59s\n- openstack-tox-docs http://logs.openstack.org/54/650054/14/check/openstack-tox-docs/8bbc068/html/ : SUCCESS in 2m 28s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/54/650054/14/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/acc8613/ : SUCCESS in 1h 03m 19s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/14/check/tripleo-ci-centos-7-standalone-os-tempest/369e4a8/ : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":14},{"id":"8161f6d4e2fafe5ae027b5278aebb8baaf09394c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 13:19:41.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"c44b63d376430ce338c5e2697a6bdf2129648dc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 14:55:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"c44c3e4971d3aacfb5a281b5140ef93ff883bbf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 14:55:17.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/54/650054/14/gate/openstack-ansible-linters/03f6fc0/ : SUCCESS in 5m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_metal-debian-stable/794147f/ : SUCCESS in 58m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_metal-centos-7/0ef57c7/ : SUCCESS in 1h 19m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_metal-opensuse-150/659f2d0/ : SUCCESS in 1h 09m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/15f2658/ : SUCCESS in 1h 04m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/a370527/ : SUCCESS in 56m 24s\n- openstack-tox-docs http://logs.openstack.org/54/650054/14/gate/openstack-tox-docs/2f14aba/html/ : SUCCESS in 2m 38s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/54/650054/14/gate/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/4dd1bc7/ : SUCCESS in 59m 58s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/54/650054/14/gate/tripleo-ci-centos-7-standalone-os-tempest/1b4ef32/ : SUCCESS in 1h 11m 15s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"6f9fd609690bfce244119b471340ef0d52ab53b5","revisions":{"f9d0ff8159daaf537a9aad0ee723ba1fc2c5ff54":{"kind":"REWORK","_number":1,"created":"2019-04-04 15:39:46.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/1"}}},"commit":{"parents":[{"commit":"8d81e1eff2e3e9fc6a202079288d68e304a2b747","subject":"Fix Generate test-list file task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8d81e1eff2e3e9fc6a202079288d68e304a2b747"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:39:37.000000000","tz":330},"subject":"Switch to import{task/role} in os_tempest","message":"Switch to import{task/role} in os_tempest\n\nSwitch to include_task to import task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f9d0ff8159daaf537a9aad0ee723ba1fc2c5ff54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f9d0ff8159daaf537a9aad0ee723ba1fc2c5ff54"}]},"branch":"refs/heads/master"},"5c4dfa48789d96557740b3f7bdade28df55d24b2":{"kind":"REWORK","_number":2,"created":"2019-04-05 12:28:44.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/2"}}},"commit":{"parents":[{"commit":"8d81e1eff2e3e9fc6a202079288d68e304a2b747","subject":"Fix Generate test-list file task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8d81e1eff2e3e9fc6a202079288d68e304a2b747"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-05 12:28:29.000000000","tz":330},"subject":"Switch to import{task/role} in os_tempest","message":"Switch to import{task/role} in os_tempest\n\nSwitch to include_task to import task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5c4dfa48789d96557740b3f7bdade28df55d24b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5c4dfa48789d96557740b3f7bdade28df55d24b2"}]},"branch":"refs/heads/master"},"733d4ae59165f414d71fb274a78edb1e3599629a":{"kind":"REWORK","_number":3,"created":"2019-04-08 03:28:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/3"}}},"commit":{"parents":[{"commit":"8d81e1eff2e3e9fc6a202079288d68e304a2b747","subject":"Fix Generate test-list file task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8d81e1eff2e3e9fc6a202079288d68e304a2b747"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-08 03:28:18.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/733d4ae59165f414d71fb274a78edb1e3599629a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/733d4ae59165f414d71fb274a78edb1e3599629a"}]},"branch":"refs/heads/master"},"f951292c91c82b272a9aebf09696e585c1364917":{"kind":"REWORK","_number":4,"created":"2019-04-08 09:28:07.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/4"}}},"commit":{"parents":[{"commit":"8d81e1eff2e3e9fc6a202079288d68e304a2b747","subject":"Fix Generate test-list file task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8d81e1eff2e3e9fc6a202079288d68e304a2b747"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-08 09:27:53.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f951292c91c82b272a9aebf09696e585c1364917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f951292c91c82b272a9aebf09696e585c1364917"}]},"branch":"refs/heads/master"},"e19d1532018cc35079207dea06f011b9acc4e960":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-04-10 13:43:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/5"}}},"commit":{"parents":[{"commit":"49f992bf57b91e14d2759c3fba890b218f32798b","subject":"Merge \"debian: add ci and update meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/49f992bf57b91e14d2759c3fba890b218f32798b"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-04-10 13:43:27.000000000","tz":0},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e19d1532018cc35079207dea06f011b9acc4e960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e19d1532018cc35079207dea06f011b9acc4e960"}]},"branch":"refs/heads/master"},"a2b1a8c9f058013713620a55207fd70b88cef489":{"kind":"REWORK","_number":6,"created":"2019-04-11 08:01:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/6"}}},"commit":{"parents":[{"commit":"8d81e1eff2e3e9fc6a202079288d68e304a2b747","subject":"Fix Generate test-list file task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8d81e1eff2e3e9fc6a202079288d68e304a2b747"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-11 08:00:04.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/a2b1a8c9f058013713620a55207fd70b88cef489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/a2b1a8c9f058013713620a55207fd70b88cef489"}]},"branch":"refs/heads/master"},"e286186d8e2956d0c8acc1379a4cff076e52d865":{"kind":"REWORK","_number":7,"created":"2019-04-11 08:02:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/7"}}},"commit":{"parents":[{"commit":"8d81e1eff2e3e9fc6a202079288d68e304a2b747","subject":"Fix Generate test-list file task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8d81e1eff2e3e9fc6a202079288d68e304a2b747"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-11 08:02:21.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e286186d8e2956d0c8acc1379a4cff076e52d865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e286186d8e2956d0c8acc1379a4cff076e52d865"}]},"branch":"refs/heads/master"},"fac3b0428837668869e8fec3e412db7a402db277":{"kind":"REWORK","_number":8,"created":"2019-04-12 09:27:37.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/8"}}},"commit":{"parents":[{"commit":"8d81e1eff2e3e9fc6a202079288d68e304a2b747","subject":"Fix Generate test-list file task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8d81e1eff2e3e9fc6a202079288d68e304a2b747"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-12 09:27:26.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/fac3b0428837668869e8fec3e412db7a402db277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/fac3b0428837668869e8fec3e412db7a402db277"}]},"branch":"refs/heads/master"},"618e233e37cf56652845383301d09ec92ea5ee03":{"kind":"REWORK","_number":9,"created":"2019-04-15 12:56:43.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/9"}}},"commit":{"parents":[{"commit":"8d81e1eff2e3e9fc6a202079288d68e304a2b747","subject":"Fix Generate test-list file task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8d81e1eff2e3e9fc6a202079288d68e304a2b747"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-15 12:56:31.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/618e233e37cf56652845383301d09ec92ea5ee03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/618e233e37cf56652845383301d09ec92ea5ee03"}]},"branch":"refs/heads/master"},"8a6121071c1c88f30afdabcfc1b73d10dfe65305":{"kind":"REWORK","_number":10,"created":"2019-06-04 11:58:18.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/10"}}},"commit":{"parents":[{"commit":"035c0d473ea9283bda771f5f56895e7ef3b42e72","subject":"Merge \"Update cirros image to 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/035c0d473ea9283bda771f5f56895e7ef3b42e72"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-04 11:57:55.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8a6121071c1c88f30afdabcfc1b73d10dfe65305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8a6121071c1c88f30afdabcfc1b73d10dfe65305"}]},"branch":"refs/heads/master"},"f388dd0977584c7d8e717eec1443fbeab74375f6":{"kind":"REWORK","_number":11,"created":"2019-06-10 10:06:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/11"}}},"commit":{"parents":[{"commit":"035c0d473ea9283bda771f5f56895e7ef3b42e72","subject":"Merge \"Update cirros image to 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/035c0d473ea9283bda771f5f56895e7ef3b42e72"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-10 10:05:44.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f388dd0977584c7d8e717eec1443fbeab74375f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f388dd0977584c7d8e717eec1443fbeab74375f6"}]},"branch":"refs/heads/master"},"3f65c7e7d1b23b7837c76074634fee021cead03f":{"kind":"REWORK","_number":12,"created":"2019-06-11 06:00:05.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/12"}}},"commit":{"parents":[{"commit":"035c0d473ea9283bda771f5f56895e7ef3b42e72","subject":"Merge \"Update cirros image to 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/035c0d473ea9283bda771f5f56895e7ef3b42e72"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-11 05:59:47.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/3f65c7e7d1b23b7837c76074634fee021cead03f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/3f65c7e7d1b23b7837c76074634fee021cead03f"}]},"branch":"refs/heads/master"},"5e7363735fd301c425f88ce36534b9b782d3aedc":{"kind":"REWORK","_number":13,"created":"2019-06-11 10:23:12.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/13"}}},"commit":{"parents":[{"commit":"035c0d473ea9283bda771f5f56895e7ef3b42e72","subject":"Merge \"Update cirros image to 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/035c0d473ea9283bda771f5f56895e7ef3b42e72"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-11 10:22:56.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5e7363735fd301c425f88ce36534b9b782d3aedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5e7363735fd301c425f88ce36534b9b782d3aedc"}]},"branch":"refs/heads/master"},"6f9fd609690bfce244119b471340ef0d52ab53b5":{"kind":"REWORK","_number":14,"created":"2019-06-12 06:33:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/54/650054/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/650054/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/650054/14"}}},"commit":{"parents":[{"commit":"035c0d473ea9283bda771f5f56895e7ef3b42e72","subject":"Merge \"Update cirros image to 0.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/035c0d473ea9283bda771f5f56895e7ef3b42e72"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-04 15:37:52.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-12 06:32:59.000000000","tz":330},"subject":"Switch to import_task in os_tempest","message":"Switch to import_task in os_tempest\n\nSwitch to include_task to import_task and include_role to import\nrole in the role.\n\nChange-Id: I6b2d2ceead9cc16afc5c7c141c400b8ee61a9da5\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f9fd609690bfce244119b471340ef0d52ab53b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f9fd609690bfce244119b471340ef0d52ab53b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
