)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~651896","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I47543778487b6a9a2972b29d270d199a4d1e9b61","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"calico-tests","hashtags":[],"change_id":"I47543778487b6a9a2972b29d270d199a4d1e9b61","subject":"Do not ping router when public net is local","status":"MERGED","created":"2019-04-11 18:05:12.000000000","updated":"2019-05-09 20:24:28.000000000","submitted":"2019-05-09 20:24:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651896-1557433468672-d3510572","meta_rev_id":"ccd95b7677f7a13a2e1d3ad11ebdf575f52ce63a","_number":651896,"virtual_id_number":651896,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-09 20:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-09 18:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-09 07:15:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":2,"date":"2019-05-09 13:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"date":"2019-05-01 15:59:01.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-05-09 18:50:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 07:15:41.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-05-09 13:04:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-09 18:50:20.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-09 20:24:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cc49be1ed196609ff8e2d9e4adbdd2c5300012f","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-04-11 18:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7ebf0068e283d11deeff5705b49b5b60449e39a","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-04-11 18:07:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f5abcbb7afa3022a0f63e85329e2a0f228087ae8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 19:52:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/96/651896/1/check/openstack-ansible-linters/c21ad6c/ : SUCCESS in 11m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/96/651896/1/check/openstack-ansible-deploy-aio_metal-debian-stable/92a4ca0/ : SUCCESS in 1h 15m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/96/651896/1/check/openstack-ansible-deploy-aio_metal-centos-7/f7cacc7/ : SUCCESS in 1h 15m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/96/651896/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d5947a3/ : SUCCESS in 1h 03m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/96/651896/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e781705/ : SUCCESS in 1h 00m 39s\n- openstack-tox-docs http://logs.openstack.org/96/651896/1/check/openstack-tox-docs/782b35b/html/ : SUCCESS in 4m 43s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/96/651896/1/check/openstack-ansible-functional-tempestconf-centos-7/9b3c8b4/ : SUCCESS in 1h 28m 56s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/96/651896/1/check/tripleo-ci-centos-7-standalone-os-tempest/99c53d2/ : SUCCESS in 1h 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ed3b3f38556cb87141fe08fd10eac3640f460461","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-01 15:59:01.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"ca220033172f7a01fd1b91ec39d7ab3caa0c18fd","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-05-09 07:15:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cbca138f031a9250a1c244fdeba946820571823f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-09 13:04:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a8e8957b5b94177eeaa00c73ea677c11275c348","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-09 18:50:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c7ee17ecb28e7659b3a8268f7a30b4a7ac41926","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:50:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2056a146f9564a57b5c8b0141b849cc63eb18a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 20:24:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/96/651896/1/gate/openstack-ansible-linters/54c1688/ : SUCCESS in 5m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/96/651896/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/6c5f27a/ : SUCCESS in 1h 00m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/96/651896/1/gate/openstack-ansible-deploy-aio_metal-centos-7/550245a/ : SUCCESS in 1h 27m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/96/651896/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/a923d82/ : SUCCESS in 1h 10m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/96/651896/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ff95157/ : SUCCESS in 57m 46s\n- openstack-tox-docs http://logs.openstack.org/96/651896/1/gate/openstack-tox-docs/b06a0e0/html/ : SUCCESS in 2m 52s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/96/651896/1/gate/openstack-ansible-functional-tempestconf-centos-7/83e2d17/ : SUCCESS in 1h 08m 53s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/96/651896/1/gate/tripleo-ci-centos-7-standalone-os-tempest/0529319/ : SUCCESS in 1h 07m 41s","accounts_in_message":[],"_revision_number":1},{"id":"20e0b9b52825166de20e3a4cf36f3fe526608bd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 20:24:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"897f80a37feb5e5b7caf2d2daf13ebd5f2e80688","revisions":{"897f80a37feb5e5b7caf2d2daf13ebd5f2e80688":{"kind":"REWORK","_number":1,"created":"2019-04-11 18:05:12.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/96/651896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/96/651896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/96/651896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/96/651896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/96/651896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/96/651896/1"}}},"commit":{"parents":[{"commit":"49f992bf57b91e14d2759c3fba890b218f32798b","subject":"Merge \"debian: add ci and update meta\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/49f992bf57b91e14d2759c3fba890b218f32798b"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-04-11 18:02:29.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-04-11 18:02:29.000000000","tz":-300},"subject":"Do not ping router when public net is local","message":"Do not ping router when public net is local\n\nDon\u0027t bother trying to ping the router address when the public\nnetwork type is \u0027local\u0027. In calico, a local network type is used,\nand the router is not set up on the host until an instance using\nthat network is created. Since this check occurs before the\ntempest run (and before any instances are created), the gateway\nis not expected to ping.\n\nChange-Id: I47543778487b6a9a2972b29d270d199a4d1e9b61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/897f80a37feb5e5b7caf2d2daf13ebd5f2e80688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/897f80a37feb5e5b7caf2d2daf13ebd5f2e80688"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
