)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~657778","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I7600f1750538335c1348e7d042e6bf450910e12c","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"fix-osa-gates","hashtags":[],"change_id":"I7600f1750538335c1348e7d042e6bf450910e12c","subject":"Remove install_test_requirements flag","status":"MERGED","created":"2019-05-08 11:13:39.000000000","updated":"2019-05-08 18:30:30.000000000","submitted":"2019-05-08 18:30:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657778-1557340230202-12e70c14","meta_rev_id":"574ac970c39d9e99b519604229d8788cf2691604","_number":657778,"virtual_id_number":657778,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-08 18:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-08 18:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-08 16:36:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-05-08 15:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-08 18:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 11:43:53.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-05-08 15:29:46.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-08 16:36:04.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-08 18:26:01.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-08 18:30:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c548967af251704621f44fb27ec1fbcc6764aef","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-08 11:13:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"763737a6ce2494aa2cb44f66c6d567ff634b329f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-08 11:37:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"129b82db6b8586ceca7f2525ba4d6e1b73de162a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-08 11:40:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d42774af27a8c90b833a42b9c74a81ae9699c1a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-08 11:43:35.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nRemoved those stuffs","accounts_in_message":[],"_revision_number":1},{"id":"0569d1bda81bd38c9a1878f137dea9cb9fb1994e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 12:56:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/657778/2/check/openstack-ansible-linters/ae6c01e/ : SUCCESS in 5m 48s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/78/657778/2/check/openstack-ansible-deploy-aio_metal-debian-stable/59321fd/ : SUCCESS in 57m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/78/657778/2/check/openstack-ansible-deploy-aio_metal-centos-7/8e0e1c3/ : SUCCESS in 1h 10m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/78/657778/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/6ec1d61/ : SUCCESS in 1h 02m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/78/657778/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/aa336bc/ : FAILURE in 51m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/78/657778/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/0dd9ce4/ : FAILURE in 44m 54s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/78/657778/2/check/openstack-tox-docs/d1c5236/html/ : SUCCESS in 2m 08s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/78/657778/2/check/openstack-ansible-functional-tempestconf-centos-7/9891e4e/ : FAILURE in 53m 34s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/78/657778/2/check/tripleo-ci-centos-7-standalone-os-tempest/c990fc8/ : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a6357ffedfe7f1370ae602aae099f4bf9825485c","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-08 15:05:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d95c9e4ec95360bbb3a612b1379ffaaed918d722","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-08 15:14:28.000000000","message":"Topic changed from fix to fix-osa-gates","accounts_in_message":[],"_revision_number":3},{"id":"592ed2503ab513148e09def00a0e9ac8a1deb26b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-08 15:29:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"591780b37d5569b11c71d301c2222e1652781327","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-08 16:06:55.000000000","message":"Patch Set 3: Code-Review+1\n\nwaiting for ci","accounts_in_message":[],"_revision_number":3},{"id":"d905e2ea34136deebe45a9de1a1fbb1952ca9791","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 16:33:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/657778/3/check/openstack-ansible-linters/85299c4/ : SUCCESS in 5m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/78/657778/3/check/openstack-ansible-deploy-aio_metal-debian-stable/215f820/ : SUCCESS in 58m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/78/657778/3/check/openstack-ansible-deploy-aio_metal-centos-7/5a3b425/ : SUCCESS in 1h 22m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/78/657778/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/08d58a6/ : SUCCESS in 1h 01m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/78/657778/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/640b83f/ : FAILURE in 57m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/78/657778/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/641a7ae/ : FAILURE in 37m 29s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/78/657778/3/check/openstack-tox-docs/282445f/html/ : SUCCESS in 2m 40s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/78/657778/3/check/openstack-ansible-functional-tempestconf-centos-7/44a6709/ : SUCCESS in 1h 06m 41s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/78/657778/3/check/tripleo-ci-centos-7-standalone-os-tempest/a3d9cc5/ : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":3},{"id":"e3c0fd356fea97a632b43835cd6d5fb83a7a0048","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-08 16:36:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ecb63b56703f0d1242a044f9b6d75d9dd9f7accb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 17:06:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aa802bda77806db4dff5d92df072dbfbbfc298b4","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-08 17:19:14.000000000","message":"Patch Set 3: Code-Review-1\n\nThis doesn\u0027t work yet.\n\nLook at results here:\nhttps://review.opendev.org/#/c/657834/","accounts_in_message":[],"_revision_number":3},{"id":"2e458af3bcb14e627839ff97c196593218a234e9","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-08 17:21:45.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"5f50efae9c2c1a141e0583a567afdc750d418bdd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-08 17:26:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b954a4082daf7a0321d48168cb147d9af0bb6881","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-08 17:38:22.000000000","message":"Patch Set 3: Code-Review+1\n\nTest failure was due to Zuul cloned roles not being used due to https://review.opendev.org/#/c/656599/ being unmerged.\n\nThis patch LGTM but waiting to see a CI pass on https://review.opendev.org/#/c/657834 for openstack-ansible-deploy-aio_metal-ubuntu-bionic since that is what this patch intends to fix.","accounts_in_message":[],"_revision_number":3},{"id":"fe6e33d9dddcf5fd2989496c8153f08e6fc9c785","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-08 18:26:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"60ab3d0d4c977d785557e92dd9f29fadd717f5e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 18:30:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/657778/3/gate/openstack-ansible-linters/4322250/ : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/78/657778/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/25b3a22/ : SUCCESS in 1h 00m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/78/657778/3/gate/openstack-ansible-deploy-aio_metal-centos-7/5e24c8f/ : SUCCESS in 1h 19m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/78/657778/3/gate/openstack-ansible-deploy-aio_metal-opensuse-150/33b0aa6/ : SUCCESS in 1h 02m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/78/657778/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dc0e620/ : FAILURE in 48m 10s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/78/657778/3/gate/openstack-tox-docs/a70a75a/html/ : SUCCESS in 2m 22s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/78/657778/3/gate/openstack-ansible-functional-tempestconf-centos-7/aeab202/ : SUCCESS in 1h 09m 09s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/78/657778/3/gate/tripleo-ci-centos-7-standalone-os-tempest/53a4a0b/ : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":3},{"id":"574ac970c39d9e99b519604229d8788cf2691604","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 18:30:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6af0e09b3a5cc256484e71644d466fc7fb30fd6f","revisions":{"5dd2b7418b5c7d477a03e0195810feb138b9fecb":{"kind":"REWORK","_number":1,"created":"2019-05-08 11:13:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/78/657778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/78/657778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/1"}}},"commit":{"parents":[{"commit":"fbc1d742d2650a9ccc37a3a4ab05baa47d26d1da","subject":"Merge \"missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/fbc1d742d2650a9ccc37a3a4ab05baa47d26d1da"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-08 11:11:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-08 11:13:36.000000000","tz":330},"subject":"Remove install_test_requirements flag","message":"Remove install_test_requirements flag\n\nDue to tempest plugin split, all the tempest plugin requirements\nare kept in requirements.txt, not in test_requirements.txt, so\nit does not make any sense to install it.\n\nSo removing the flag from the role.\n\nChange-Id: I7600f1750538335c1348e7d042e6bf450910e12c\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5dd2b7418b5c7d477a03e0195810feb138b9fecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5dd2b7418b5c7d477a03e0195810feb138b9fecb"}]},"branch":"refs/heads/master"},"21035b126eb49bfd03360ec769c91cd65cf3c79a":{"kind":"REWORK","_number":2,"created":"2019-05-08 11:40:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/78/657778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/78/657778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/2"}}},"commit":{"parents":[{"commit":"fbc1d742d2650a9ccc37a3a4ab05baa47d26d1da","subject":"Merge \"missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/fbc1d742d2650a9ccc37a3a4ab05baa47d26d1da"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-08 11:11:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-08 11:40:48.000000000","tz":330},"subject":"Remove install_test_requirements flag","message":"Remove install_test_requirements flag\n\nDue to tempest plugin split, all the tempest plugin requirements\nare kept in requirements.txt, not in test_requirements.txt, so\nit does not make any sense to install it.\n\nSo removing the flag from the role.\n\nChange-Id: I7600f1750538335c1348e7d042e6bf450910e12c\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/21035b126eb49bfd03360ec769c91cd65cf3c79a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/21035b126eb49bfd03360ec769c91cd65cf3c79a"}]},"branch":"refs/heads/master"},"6af0e09b3a5cc256484e71644d466fc7fb30fd6f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-08 15:05:06.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/78/657778/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/78/657778/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/78/657778/3"}}},"commit":{"parents":[{"commit":"ebdff30fd8cddfd24f9a945db7fbd9420ef5bdad","subject":"Temporarily set bionic job to non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/ebdff30fd8cddfd24f9a945db7fbd9420ef5bdad"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-08 11:11:04.000000000","tz":330},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-05-08 15:05:06.000000000","tz":0},"subject":"Remove install_test_requirements flag","message":"Remove install_test_requirements flag\n\nDue to tempest plugin split, all the tempest plugin requirements\nare kept in requirements.txt, not in test_requirements.txt, so\nit does not make any sense to install it.\n\nSo removing the flag from the role.\n\nChange-Id: I7600f1750538335c1348e7d042e6bf450910e12c\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6af0e09b3a5cc256484e71644d466fc7fb30fd6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6af0e09b3a5cc256484e71644d466fc7fb30fd6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
