)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~658313","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~stable%2Fstein~Ifd4ccd9faa16ce14e7bb826197d48b6a28e50211","project":"openstack/openstack-ansible-os_tempest","branch":"stable/stein","topic":"fix-osa-gates-stable/stein","hashtags":[],"change_id":"Ifd4ccd9faa16ce14e7bb826197d48b6a28e50211","subject":"Revert \"Temporarily set bionic \u0026 centos_distro jobs to non-voting\"","status":"MERGED","created":"2019-05-10 09:01:10.000000000","updated":"2019-05-12 17:26:09.000000000","submitted":"2019-05-12 17:26:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658313-1557681969149-f41f2753","meta_rev_id":"5e5c1f2c522d9867d7828a7982c44aefa1dcb757","_number":658313,"virtual_id_number":658313,"owner":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-10 17:44:36.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-12 17:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-05-12 16:01:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-05-12 10:06:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2019-05-12 16:01:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 10:06:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-12 16:01:54.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-12 17:26:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6346e068feeb690d5ac82ff2efd95485b5c82d83","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 09:01:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09a1f552c70fec0cf2620c530b006e47dec5caff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 10:53:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/13/658313/1/check/openstack-ansible-linters/c350698/ : SUCCESS in 7m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_metal-debian-stable/1ff47ec/ : SUCCESS in 1h 49m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_metal-centos-7/02d28bb/ : SUCCESS in 1h 21m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/767b83c/ : FAILURE in 55m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/932ba27/ : FAILURE in 44m 40s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/13/658313/1/check/openstack-tox-docs/150c3e3/html/ : SUCCESS in 2m 36s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/13/658313/1/check/openstack-ansible-functional-tempestconf-centos-7/8903be6/ : FAILURE in 48m 58s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/13/658313/1/check/tripleo-ci-centos-7-standalone-os-tempest/5c8ba39/ : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b4d2a01a60adfff00337eee09237544f46eecc0e","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 12:16:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7ceac2652dca3d5c380a5c53d8a757ba399e506","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 13:55:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/13/658313/1/check/openstack-ansible-linters/df7ff89/ : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_metal-debian-stable/06af181/ : SUCCESS in 58m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_metal-centos-7/db07989/ : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fbc3ce3/ : FAILURE in 46m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/b22d170/ : FAILURE in 37m 29s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/13/658313/1/check/openstack-tox-docs/d49d927/html/ : SUCCESS in 2m 23s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/13/658313/1/check/openstack-ansible-functional-tempestconf-centos-7/7374861/ : SUCCESS in 1h 05m 09s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/13/658313/1/check/tripleo-ci-centos-7-standalone-os-tempest/bca499b/ : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":1},{"id":"57dc6936f7652fedfb570c5a84f86a5285aaf485","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 17:44:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b8bc3cfee631a5e70210f70fdadf14e1005d57a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:10:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/658313/1/check/openstack-ansible-linters/cab2dba/ : SUCCESS in 4m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_metal-debian-stable/14a0bc5/ : SUCCESS in 51m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_metal-centos-7/3f07c60/ : SUCCESS in 1h 17m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f177e25/ : SUCCESS in 50m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/13/658313/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/4315897/ : FAILURE in 40m 57s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/13/658313/1/check/openstack-tox-docs/ebbbdad/html/ : SUCCESS in 2m 45s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/13/658313/1/check/openstack-ansible-functional-tempestconf-centos-7/0c31b26/ : SUCCESS in 1h 03m 49s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/13/658313/1/check/tripleo-ci-centos-7-standalone-os-tempest/6913b8a/ : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"48944ff3f8545aa2db3709e3f15db8e3f0c1c021","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-12 10:06:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c8b46d8c7beb7ffe916b502776a354fbb16bdd7","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-12 16:01:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5a308de48bada5fd89715797f7c4686a335782e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 16:02:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1503a673179af1389c23537fef7c10e41f820340","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:26:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/658313/1/gate/openstack-ansible-linters/f23b02b/ : SUCCESS in 4m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/13/658313/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/ef5ef75/ : SUCCESS in 55m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/13/658313/1/gate/openstack-ansible-deploy-aio_metal-centos-7/c175c16/ : SUCCESS in 1h 09m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/13/658313/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/045a7ce/ : SUCCESS in 50m 25s\n- openstack-tox-docs http://logs.openstack.org/13/658313/1/gate/openstack-tox-docs/610ec93/html/ : SUCCESS in 3m 10s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/13/658313/1/gate/openstack-ansible-functional-tempestconf-centos-7/38f5d0b/ : SUCCESS in 1h 05m 34s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/13/658313/1/gate/tripleo-ci-centos-7-standalone-os-tempest/f5b7878/ : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d36ae0a9d56bece7095fb717f6530c5dfb650e5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:26:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b5762c50e981765fda1a24d385b632be629333b","revisions":{"4b5762c50e981765fda1a24d385b632be629333b":{"kind":"REWORK","_number":1,"created":"2019-05-10 09:01:10.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/13/658313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/13/658313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/13/658313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/13/658313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/13/658313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/13/658313/1"}}},"commit":{"parents":[{"commit":"91c2a790106e1a6f55d99b7fb962b11dd79bc6ba","subject":"Remove install_test_requirements flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/91c2a790106e1a6f55d99b7fb962b11dd79bc6ba"}]}],"author":{"name":"Jesse Pretorius","email":"jesse@odyssey4.me","date":"2019-05-10 08:59:03.000000000","tz":60},"committer":{"name":"Jesse Pretorius","email":"jesse@odyssey4.me","date":"2019-05-10 08:59:33.000000000","tz":60},"subject":"Revert \"Temporarily set bionic \u0026 centos_distro jobs to non-voting\"","message":"Revert \"Temporarily set bionic \u0026 centos_distro jobs to non-voting\"\n\nThis reverts commit c19bddfad7e6c80824d62d760634f12c74234749.\n\nDepends-On: https://review.opendev.org/658289\nChange-Id: Ifd4ccd9faa16ce14e7bb826197d48b6a28e50211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/4b5762c50e981765fda1a24d385b632be629333b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/4b5762c50e981765fda1a24d385b632be629333b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
