)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~665595","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~stable%2Fstein~I3cdfe48cae28969dea381bfa325c4bb4769fb4d1","project":"openstack/openstack-ansible-os_tempest","branch":"stable/stein","hashtags":[],"change_id":"I3cdfe48cae28969dea381bfa325c4bb4769fb4d1","subject":"Set default tempest_install_method for meta handlers","status":"MERGED","created":"2019-06-17 06:59:28.000000000","updated":"2019-06-18 17:31:01.000000000","submitted":"2019-06-18 17:31:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665595-1560879061395-d3b36feb","meta_rev_id":"f2c8f4a376a0a0d70dfc2ef579763704cc484a04","_number":665595,"virtual_id_number":665595,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-18 17:31:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-06-18 16:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-18 16:07:58.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-17 23:11:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":2,"date":"2019-06-17 19:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-06-18 16:07:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 19:43:04.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-17 23:11:21.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-06-18 16:07:50.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-18 16:07:58.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-18 17:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"840c6eba27e30abac7373ef905a890c1818c5ca5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-17 06:59:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdfebfd79b0c2c26bcadf81228da4aaa442d23b1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-17 07:01:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f2941f1498c9f7e326005caaa8c881beea4b3ec4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 08:54:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/95/665595/2/check/openstack-ansible-linters/4b56365/ : SUCCESS in 8m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/95/665595/2/check/openstack-ansible-deploy-aio_metal-debian-stable/82be461/ : SUCCESS in 58m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/95/665595/2/check/openstack-ansible-deploy-aio_metal-centos-7/4e16b40/ : SUCCESS in 1h 18m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/95/665595/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/08633f4/ : SUCCESS in 53m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/95/665595/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8f2c0d6/ : FAILURE in 43m 06s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/95/665595/2/check/openstack-tox-docs/3c45c35/html/ : SUCCESS in 2m 39s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/95/665595/2/check/openstack-ansible-functional-tempestconf-centos-7/0d67e82/ : SUCCESS in 1h 07m 26s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/95/665595/2/check/tripleo-ci-centos-7-standalone-os-tempest/82320fd/ : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b6e707272c6be1937832f24086ca66f6820387ff","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-17 19:43:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cccdf428e7946205454f7679537ce0806b836ff5","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-06-17 23:11:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b0ca00701d801b596d94b95d24d17c094d2eaa7f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-18 16:07:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ea1c00512935e995ddaa4079a5e07f76a6721061","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-18 16:07:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c778c06c8478f20ac8a1c1e28650ed74fff9302f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:08:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"58f137191261b4b6d73ee1598c96505b77b54942","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 17:31:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f2c8f4a376a0a0d70dfc2ef579763704cc484a04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 17:31:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/95/665595/2/gate/openstack-ansible-linters/70da66e/ : SUCCESS in 8m 26s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/95/665595/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/6a41e0c/ : SUCCESS in 59m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/95/665595/2/gate/openstack-ansible-deploy-aio_metal-centos-7/8cff5af/ : SUCCESS in 1h 14m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/95/665595/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/4c2d97a/ : SUCCESS in 1h 02m 34s\n- openstack-tox-docs http://logs.openstack.org/95/665595/2/gate/openstack-tox-docs/0421667/html/ : SUCCESS in 4m 04s\n- openstack-ansible-functional-tempestconf-centos-7 http://logs.openstack.org/95/665595/2/gate/openstack-ansible-functional-tempestconf-centos-7/fc11311/ : SUCCESS in 1h 14m 19s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/95/665595/2/gate/tripleo-ci-centos-7-standalone-os-tempest/84c6143/ : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b9e97cb4873c5036b86e28e79b61555a698edf29","revisions":{"c8261851e15045ca81fe733637ad9fa90efff50b":{"kind":"REWORK","_number":1,"created":"2019-06-17 06:59:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/95/665595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/95/665595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/95/665595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/95/665595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/95/665595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/95/665595/1"}}},"commit":{"parents":[{"commit":"6b16c973e2e77304ad362dd7da6dbe9cb85ba6a4","subject":"Do not ping router when public net is local","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6b16c973e2e77304ad362dd7da6dbe9cb85ba6a4"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-13 21:51:13.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-17 06:59:04.000000000","tz":180},"subject":"Set default tempest_install_method for meta handlers","message":"Set default tempest_install_method for meta handlers\n\nFor non-integrated tests tempest_install_method is not defined,\nso when python_venv_build tries to launch meta handler, it fails with\nundefined variable. Once default is set, this will prevent problems\n\nDepends-On: https://review.opendev.org/665489/\n\nChange-Id: I3cdfe48cae28969dea381bfa325c4bb4769fb4d1\n(cherry picked from commit 7e6d40ccf15d870fdfd59636081603d330476015)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/c8261851e15045ca81fe733637ad9fa90efff50b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/c8261851e15045ca81fe733637ad9fa90efff50b"}]},"branch":"refs/heads/stable/stein"},"b9e97cb4873c5036b86e28e79b61555a698edf29":{"kind":"REWORK","_number":2,"created":"2019-06-17 07:01:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/95/665595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/95/665595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/95/665595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/95/665595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/95/665595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/95/665595/2"}}},"commit":{"parents":[{"commit":"6b16c973e2e77304ad362dd7da6dbe9cb85ba6a4","subject":"Do not ping router when public net is local","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6b16c973e2e77304ad362dd7da6dbe9cb85ba6a4"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-13 21:51:13.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-17 07:01:00.000000000","tz":180},"subject":"Set default tempest_install_method for meta handlers","message":"Set default tempest_install_method for meta handlers\n\nFor non-integrated tests tempest_install_method is not defined,\nso when python_venv_build tries to launch meta handler, it fails with\nundefined variable. Once default is set, this will prevent problems\n\nDepends-On: https://review.opendev.org/665489/\n\nChange-Id: I3cdfe48cae28969dea381bfa325c4bb4769fb4d1\n(cherry picked from commit 7e6d40ccf15d870fdfd59636081603d330476015)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/b9e97cb4873c5036b86e28e79b61555a698edf29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/b9e97cb4873c5036b86e28e79b61555a698edf29"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
