)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~672225","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I4dc81706fd1ba39097ec3a70648897a32083f173","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"stestr-3","hashtags":[],"change_id":"I4dc81706fd1ba39097ec3a70648897a32083f173","subject":"Update stestr_executable variable for RHEL-8","status":"MERGED","created":"2019-07-23 08:30:32.000000000","updated":"2019-07-24 15:44:33.000000000","submitted":"2019-07-24 15:44:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672225-1563983073762-cd2898e8","meta_rev_id":"389746a9383207cb42e51ac3302a36857be064b9","_number":672225,"virtual_id_number":672225,"owner":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-24 15:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-24 12:38:34.000000000","_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"recommended":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":1,"date":"2019-07-23 14:10:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-07-23 16:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2019-07-23 16:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"date":"2019-07-23 14:11:00.000000000","_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2019-07-23 16:47:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"date":"2019-07-24 10:28:24.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 14:10:48.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-07-23 14:11:00.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-07-23 16:04:28.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-23 16:47:11.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-24 10:28:24.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-24 15:44:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21508568dd537e7eb48fd1ce00d952511c6a49f2","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-07-23 08:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"326a6e78ff6bd6ffa64081d0bbffd5c07c5b4ae3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 10:18:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/25/672225/1/check/openstack-ansible-linters/447d2e4/ : SUCCESS in 7m 47s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-debian-stable/d2446fb/ : SUCCESS in 57m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-centos-7/60c7671/ : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/a4cc46d/ : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/94f541e/ : SUCCESS in 59m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/55e8ea8/ : SUCCESS in 1h 06m 30s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/7436beb/ : FAILURE in 45m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/d7bd948/ : SUCCESS in 45m 10s\n- openstack-tox-docs http://logs.openstack.org/25/672225/1/check/openstack-tox-docs/63c6b64/html/ : SUCCESS in 3m 59s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/a746b0b/ : SUCCESS in 59m 20s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/25/672225/1/check/tripleo-ci-centos-7-standalone-os-tempest/d190142/ : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"69f0fce6f5166a741e8a979346bce7c670e051fb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-23 14:10:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b76657d9032c41c7259f789e18e0a4d6528bfc44","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-23 16:04:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"72a61a6575ff58318b5aa44e263ae3c7b43877ac","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-23 16:47:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73c099574adfbd475c5c75fe481c7abd42ac2eb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 16:47:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04db2f59d6db8e387619e4bcec6c98e5be77511c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 18:08:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/25/672225/1/gate/openstack-ansible-linters/9d1f2e6/ : SUCCESS in 5m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/79b8cf0/ : SUCCESS in 58m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_metal-centos-7/c6df6c0/ : SUCCESS in 1h 16m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/3f1bd2b/ : FAILURE in 15m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c28f4d4/ : SUCCESS in 1h 00m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/d243777/ : SUCCESS in 54m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/8826cf3/ : SUCCESS in 48m 01s\n- openstack-tox-docs http://logs.openstack.org/25/672225/1/gate/openstack-tox-docs/fd9f404/html/ : SUCCESS in 3m 26s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/13e6fda/ : SUCCESS in 1h 11m 15s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/25/672225/1/gate/tripleo-ci-centos-7-standalone-os-tempest/6e0d948/ : SUCCESS in 1h 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"dee821f504185bf32630f9c6e59e8f46a87deaa6","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-23 19:39:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d0ceca9b00d5fa19c057ab96d4ac4369381ca39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 21:05:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/25/672225/1/check/openstack-ansible-linters/4a33e76/ : SUCCESS in 6m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-debian-stable/a995eea/ : SUCCESS in 1h 22m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-centos-7/879e4d4/ : SUCCESS in 1h 22m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/5ea6c29/ : RETRY_LIMIT in 2m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9968645/ : SUCCESS in 1h 07m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8e141d6/ : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/c19b2a1/ : RETRY_LIMIT in 3m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/f6ba858/ : SUCCESS in 1h 05m 32s\n- openstack-tox-docs http://logs.openstack.org/25/672225/1/check/openstack-tox-docs/ee539f6/html/ : SUCCESS in 3m 01s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/7b6b4fd/ : SUCCESS in 1h 06m 30s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/25/672225/1/check/tripleo-ci-centos-7-standalone-os-tempest/89eb59e/ : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e85345e07da3b36e371231dd6a3c791eba9d61e0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-24 04:03:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"702786717972ffbd3cc9aaf195526500cfbcbb68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 06:01:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/25/672225/1/check/openstack-ansible-linters/b6099e8/ : SUCCESS in 6m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-debian-stable/3a0dc61/ : FAILURE in 50m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-centos-7/1b20bed/ : FAILURE in 1h 07m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/ff377f8/ : RETRY_LIMIT in 2m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5fc70e9/ : SUCCESS in 1h 04m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/6acc30e/ : FAILURE in 1h 01m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/a5abf01/ : FAILURE in 56m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/bdb2d63/ : SUCCESS in 48m 45s\n- openstack-tox-docs http://logs.openstack.org/25/672225/1/check/openstack-tox-docs/9b12b57/html/ : SUCCESS in 3m 18s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/94c4efc/ : FAILURE in 47m 40s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/25/672225/1/check/tripleo-ci-centos-7-standalone-os-tempest/7743e89/ : FAILURE in 1h 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6d0ed5a766959477c3b5457d516ccb403ae180ac","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-24 06:22:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d90f080c702132599ace15e26d37392e4edc4714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 07:48:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/25/672225/1/check/openstack-ansible-linters/fa8240f/ : SUCCESS in 7m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-debian-stable/5cc27ad/ : FAILURE in 1h 00m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-centos-7/81f916a/ : FAILURE in 1h 10m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/c8e9b9f/ : FAILURE in 1h 01m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/49ef3a2/ : SUCCESS in 1h 03m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/cacc806/ : FAILURE in 55m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/9f248e1/ : FAILURE in 46m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/b8449a0/ : SUCCESS in 46m 54s\n- openstack-tox-docs http://logs.openstack.org/25/672225/1/check/openstack-tox-docs/2efe587/html/ : SUCCESS in 3m 31s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/f179cc3/ : FAILURE in 53m 42s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/25/672225/1/check/tripleo-ci-centos-7-standalone-os-tempest/3b1c0ca/ : FAILURE in 53m 00s","accounts_in_message":[],"_revision_number":1},{"id":"41028cb49f6a6d92dfe2ecd31daf2aa44349798b","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-07-24 09:18:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab3fe8e898af10f765b1f0ddceb99c8e973a5a5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:26:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/25/672225/1/check/openstack-ansible-linters/0824e47/ : SUCCESS in 7m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-debian-stable/39c99c7/ : FAILURE in 51m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-centos-7/8d7df64/ : FAILURE in 1h 05m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/aaa7cc2/ : FAILURE in 1h 05m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ce44848/ : SUCCESS in 1h 04m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/2a2c2e0/ : FAILURE in 47m 10s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/7a0c74f/ : FAILURE in 47m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/313c2bc/ : SUCCESS in 50m 07s\n- openstack-tox-docs http://logs.openstack.org/25/672225/1/check/openstack-tox-docs/90edd00/html/ : SUCCESS in 4m 19s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/83f20a5/ : FAILURE in 54m 20s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/25/672225/1/check/tripleo-ci-centos-7-standalone-os-tempest/589a830/ : FAILURE in 1h 05m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e803e164ec64ce80c0e5b64b89047369c102a0c0","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-24 10:28:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"49c9898327518f38c384728113cb7e46cb504eb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 11:55:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/25/672225/1/check/openstack-ansible-linters/76633d1/ : SUCCESS in 6m 44s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-debian-stable/9605fee/ : FAILURE in 1h 02m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-centos-7/5920ad7/ : FAILURE in 1h 21m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/e419392/ : FAILURE in 1h 09m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/bf3b2fa/ : SUCCESS in 1h 05m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/b7429a6/ : FAILURE in 1h 01m 57s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/3efcf5c/ : FAILURE in 35m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/ee2f31d/ : SUCCESS in 54m 24s\n- openstack-tox-docs http://logs.openstack.org/25/672225/1/check/openstack-tox-docs/273b495/html/ : SUCCESS in 4m 06s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/b2f4fd1/ : FAILURE in 57m 59s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/25/672225/1/check/tripleo-ci-centos-7-standalone-os-tempest/21167b3/ : FAILURE in 1h 07m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8a1ebd430cd553735e9e8ddae5d515ada0f856df","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-24 12:13:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4f6fe9067e381d7c9b3b50e585b9f87dadf1f3e","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-07-24 12:38:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca2c2fcbad6d763fc2973de5a2695b49a6a7c629","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:51:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/25/672225/1/check/openstack-ansible-linters/9aa5dc3/ : SUCCESS in 6m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-debian-stable/a37d2e5/ : SUCCESS in 1h 16m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-centos-7/6feeb6a/ : SUCCESS in 1h 26m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/3cd7202/ : SUCCESS in 1h 21m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/46571c7/ : SUCCESS in 1h 11m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/b834d46/ : SUCCESS in 1h 01m 51s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/7b3db6d/ : FAILURE in 35m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/dc41168/ : SUCCESS in 44m 29s\n- openstack-tox-docs http://logs.openstack.org/25/672225/1/check/openstack-tox-docs/770cd51/html/ : SUCCESS in 3m 50s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/25/672225/1/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/e7b754f/ : SUCCESS in 1h 16m 27s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/25/672225/1/check/tripleo-ci-centos-7-standalone-os-tempest/6a46acb/ : SUCCESS in 1h 08m 03s","accounts_in_message":[],"_revision_number":1},{"id":"aef25cbb544497200e0d0118a389b843693ae1c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:51:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4af4ad95844e0439ea17e9f5c6c2080585206df0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 15:44:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3df9d3ea2a2e5782192d1039161a6d733df7b444","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 15:44:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/25/672225/1/gate/openstack-ansible-linters/f910383/ : SUCCESS in 7m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/ed645eb/ : SUCCESS in 1h 09m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_metal-centos-7/6a3932b/ : SUCCESS in 1h 50m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/6dfb60c/ : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2caaefc/ : SUCCESS in 56m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/8bcf03e/ : SUCCESS in 1h 22m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/d5e65c2/ : SUCCESS in 48m 51s\n- openstack-tox-docs http://logs.openstack.org/25/672225/1/gate/openstack-tox-docs/a78838f/html/ : SUCCESS in 3m 48s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/25/672225/1/gate/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/8c04e0b/ : SUCCESS in 1h 08m 02s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/25/672225/1/gate/tripleo-ci-centos-7-standalone-os-tempest/a61160e/ : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5825392971db66cc09b793e9d350dde7600f39c6","revisions":{"5825392971db66cc09b793e9d350dde7600f39c6":{"kind":"REWORK","_number":1,"created":"2019-07-23 08:30:32.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/25/672225/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/25/672225/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/25/672225/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/25/672225/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/25/672225/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/25/672225/1"}}},"commit":{"parents":[{"commit":"95ed7b840d00a4331b6c5704680d1a1953c6ba58","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/95ed7b840d00a4331b6c5704680d1a1953c6ba58"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2019-07-23 08:29:50.000000000","tz":120},"committer":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2019-07-23 08:30:11.000000000","tz":120},"subject":"Update stestr_executable variable for RHEL-8","message":"Update stestr_executable variable for RHEL-8\n\nRHEL-8 comes only with python3 and so, the executable used by stestr is\ncalled stestr-3.\n\nChange-Id: I4dc81706fd1ba39097ec3a70648897a32083f173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5825392971db66cc09b793e9d350dde7600f39c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5825392971db66cc09b793e9d350dde7600f39c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
