)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~673780","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I85c8bd602246a26c12e6ec91582f4207332ca8cf","project":"openstack/openstack-ansible-os_tempest","branch":"master","hashtags":[],"change_id":"I85c8bd602246a26c12e6ec91582f4207332ca8cf","subject":"Add manila test","status":"ABANDONED","created":"2019-07-31 10:23:51.000000000","updated":"2019-08-02 15:20:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"330a20eb18a062e45a18919a3b4af57f82ef57e6","_number":673780,"virtual_id_number":673780,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":1,"date":"2019-07-31 11:42:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"date":"2019-07-31 16:03:13.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":-1,"date":"2019-07-31 12:36:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"date":"2019-07-31 11:40:04.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 11:40:04.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-07-31 11:42:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-31 16:03:13.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"008fb19dceeb1325fe27a0159b92d86996e308bc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-31 10:23:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f851b3a67d52944fc065c4ba86211d0b33e9b07f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 11:42:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/80/673780/1/check/openstack-ansible-linters/59948f7/ : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/80/673780/1/check/openstack-ansible-deploy-aio_metal-debian-stable/9e75060/ : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/80/673780/1/check/openstack-ansible-deploy-aio_metal-centos-7/ed5905b/ : SUCCESS in 1h 17m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/80/673780/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/1acd066/ : SUCCESS in 1h 12m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/80/673780/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3fe149a/ : SUCCESS in 1h 03m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/80/673780/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/d884bfe/ : SUCCESS in 1h 05m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/80/673780/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/751de91/ : SUCCESS in 56m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/80/673780/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/8c5ef40/ : SUCCESS in 48m 42s\n- openstack-tox-docs http://logs.openstack.org/80/673780/1/check/openstack-tox-docs/ea23522/html/ : SUCCESS in 2m 32s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 http://logs.openstack.org/80/673780/1/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/81f7455/ : SUCCESS in 1h 02m 48s\n- tripleo-ci-centos-7-standalone-os-tempest http://logs.openstack.org/80/673780/1/check/tripleo-ci-centos-7-standalone-os-tempest/af5d744/ : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"28ae50f4eb0ae3dac73825f2ad1e823824cfaad5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-31 12:36:32.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"330a20eb18a062e45a18919a3b4af57f82ef57e6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-02 15:20:15.000000000","message":"Abandoned\n\nThis test should be included here https://opendev.org/openstack/openstack-ansible/src/branch/master/tests/roles/bootstrap-host/templates/user_variables_manila.yml.j2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3c6163d1a9a6c835d624d00405b11bc959a1e61","revisions":{"b3c6163d1a9a6c835d624d00405b11bc959a1e61":{"kind":"REWORK","_number":1,"created":"2019-07-31 10:23:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/80/673780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/80/673780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/80/673780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/80/673780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/80/673780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/80/673780/1"}}},"commit":{"parents":[{"commit":"6f5e0518dd055678b94df995ac7f841faa4359b4","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f5e0518dd055678b94df995ac7f841faa4359b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-31 10:23:33.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-31 10:23:33.000000000","tz":180},"subject":"Add manila test","message":"Add manila test\n\nWhen tempest_service_available_manila is set to true, manila test\nshould be executed.\n\nChange-Id: I85c8bd602246a26c12e6ec91582f4207332ca8cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/b3c6163d1a9a6c835d624d00405b11bc959a1e61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/b3c6163d1a9a6c835d624d00405b11bc959a1e61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
