)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~675353","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I198190d601cf8a922c3afaa69a3c0e42ab075735","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"ubuntu_distro_plugins","hashtags":[],"change_id":"I198190d601cf8a922c3afaa69a3c0e42ab075735","subject":"Add mixed setup for debian based OS","status":"MERGED","created":"2019-08-08 14:32:34.000000000","updated":"2019-08-13 17:47:53.000000000","submitted":"2019-08-13 17:47:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675353-1565718473414-28760403","meta_rev_id":"28b1c91b9fa908bde56f29fd26d12db4e716b07b","_number":675353,"virtual_id_number":675353,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-13 17:47:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-13 16:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-13 04:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":1,"date":"2019-08-13 08:36:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-13 16:02:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 04:40:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-13 08:36:45.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-08-13 16:02:46.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-13 17:47:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a346ed09a8f4c80c44482ec8114e347c397cf168","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-08 14:32:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"311b59486bfa0a28978b0c84c33a2fe50c268071","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 15:36:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/53/675353/1/check/openstack-ansible-linters/456ebb0/ : SUCCESS in 6m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/53/675353/1/check/openstack-ansible-deploy-aio_metal-debian-stable/86ee6f8/ : FAILURE in 21m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/53/675353/1/check/openstack-ansible-deploy-aio_metal-centos-7/52381ee/ : FAILURE in 25m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/53/675353/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/f3b58d8/ : FAILURE in 25m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/53/675353/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/97c34dc/ : FAILURE in 20m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/53/675353/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/073981b/ : FAILURE in 22m 27s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/53/675353/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/814e53f/ : FAILURE in 23m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/53/675353/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/226ab9e/ : FAILURE in 21m 33s\n- openstack-tox-docs https://logs.opendev.org/53/675353/1/check/openstack-tox-docs/e71fcf7/html/ : SUCCESS in 4m 02s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://logs.opendev.org/53/675353/1/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/2bd46ab/ : FAILURE in 22m 36s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/53/675353/1/check/tripleo-ci-centos-7-standalone-os-tempest/f230069/ : FAILURE in 46m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1ac6e23cb0c5ab4951e36352b8cef0cb89bebc17","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-08 15:53:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"345be549a62f1f2facd74dd58cde3da4140953d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 17:21:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/53/675353/2/check/openstack-ansible-linters/76a9ecd/ : SUCCESS in 7m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/53/675353/2/check/openstack-ansible-deploy-aio_metal-debian-stable/ac0a175/ : SUCCESS in 1h 06m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/53/675353/2/check/openstack-ansible-deploy-aio_metal-centos-7/99eb72d/ : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/53/675353/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/2c4f36e/ : SUCCESS in 1h 12m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/53/675353/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/bda1707/ : SUCCESS in 1h 03m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/53/675353/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/4a00562/ : SUCCESS in 59m 14s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/53/675353/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/114690c/ : SUCCESS in 52m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/53/675353/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/b6f92f9/ : SUCCESS in 45m 05s\n- openstack-tox-docs https://logs.opendev.org/53/675353/2/check/openstack-tox-docs/1d78469/html/ : SUCCESS in 2m 52s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://logs.opendev.org/53/675353/2/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/9b7317f/ : SUCCESS in 1h 14m 47s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/53/675353/2/check/tripleo-ci-centos-7-standalone-os-tempest/dca8eec/ : FAILURE in 1h 15m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d3f37a815bc22c693ec1e4e99ee2b3e5c8f87609","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-08 19:54:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b5822b5b260a376b3ccef31b74ff3366bb0a3f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-08 19:58:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"11cfb5ac7ad2868d698062f5ac73a565424b6b43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 21:27:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/53/675353/4/check/openstack-ansible-linters/4ff07db/ : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/53/675353/4/check/openstack-ansible-deploy-aio_metal-debian-stable/365835b/ : SUCCESS in 1h 01m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/53/675353/4/check/openstack-ansible-deploy-aio_metal-centos-7/b60802f/ : SUCCESS in 1h 27m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/53/675353/4/check/openstack-ansible-deploy-aio_metal-opensuse-15/02fcd3f/ : SUCCESS in 1h 13m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/53/675353/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e010c56/ : SUCCESS in 1h 01m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/53/675353/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/83ad868/ : SUCCESS in 1h 01m 29s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/53/675353/4/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/47f3924/ : SUCCESS in 56m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/53/675353/4/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/10a11e8/ : SUCCESS in 40m 58s\n- openstack-tox-docs https://logs.opendev.org/53/675353/4/check/openstack-tox-docs/2e1b6a3/html/ : SUCCESS in 4m 11s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://logs.opendev.org/53/675353/4/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/f3eef9f/ : SUCCESS in 1h 10m 33s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/53/675353/4/check/tripleo-ci-centos-7-standalone-os-tempest/296130c/ : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":4},{"id":"f856da38a53226630c63732e1caf3df5b3afda2a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-09 21:25:26.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nIt\u0027s great to have got this working for the ubuntu distro installs. I\u0027m not sure the var name tempest_mixed_setup is particularly readable - this is an alternative with two vars. In this case the var names are more descriptive of what they do and are maybe a bit clearer when used as conditionals on the tasks.\n\nAn additional conditional for including the source tasks\n\n  #the plugin install method follows the main install method except for ubuntu where there are no disto packaged plugins\n  tempest_plugin_install_method: \"{{ ((tempest_install_method \u003d\u003d \u0027distro\u0027) and (ansible_os_family | lower \u003d\u003d \u0027debian\u0027)) | ternary(\"source\", tempest_install_method) }}\"\n\nThe conditionals inside tempest_install_source.yml\n\n  #install any source packages into a venv except for plugins that come from source on an ubnutu distro install\n  tempest_source_venv_enabled: \"{{ ((tempest_install_method \u003d\u003d \u0027distro\u0027) and (ansible_os_family | lower \u003d\u003d \u0027debian\u0027)) | ternary(false, true) }}\"","accounts_in_message":[],"_revision_number":4},{"id":"ac3bb4cf98f96df6800427b35dc15c33cca534a4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-12 09:46:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1ac95e934b48048e4a54b6c1a3f539840be68bcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 11:28:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/53/675353/5/check/openstack-ansible-linters/d2b1fd6/ : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/53/675353/5/check/openstack-ansible-deploy-aio_metal-debian-stable/959c799/ : SUCCESS in 1h 09m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/53/675353/5/check/openstack-ansible-deploy-aio_metal-centos-7/a4537d9/ : SUCCESS in 1h 22m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/53/675353/5/check/openstack-ansible-deploy-aio_metal-opensuse-15/c7e3f6b/ : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/53/675353/5/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0fdc448/ : SUCCESS in 1h 06m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/53/675353/5/check/openstack-ansible-deploy-aio_distro_metal-centos-7/34c31e4/ : SUCCESS in 1h 03m 37s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/53/675353/5/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/7924dae/ : SUCCESS in 1h 02m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/53/675353/5/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/bb632e8/ : SUCCESS in 1h 31m 33s\n- openstack-tox-docs https://logs.opendev.org/53/675353/5/check/openstack-tox-docs/a1e61c6/html/ : SUCCESS in 2m 40s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://logs.opendev.org/53/675353/5/check/openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7/f10df1d/ : SUCCESS in 1h 07m 34s\n- tripleo-ci-centos-7-standalone-os-tempest https://logs.opendev.org/53/675353/5/check/tripleo-ci-centos-7-standalone-os-tempest/d575121/ : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":5},{"id":"f16f1960143c51d9be7310a6d2fdc6369073ca8b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-13 04:40:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a559b02be4d871f32d35e31572c9b745efef23d9","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-08-13 08:36:45.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"643e1fc49ad34d18f8de77f568bae079f5063c21","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-13 16:02:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0c27ecf247cd55ed4dc4ad9785054d4b1f92ab80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 16:02:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"27b9bc6fbc7a296923aa4583c6b680b579936578","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 17:47:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"6b77c7e75b2f1788fa8bd723fc3620a29676fb39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 17:47:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6d605b06759a4da5b3b46284d9c3cadc : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5d8d282054d94bdb98d004f08cfc16f4 : SUCCESS in 1h 05m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/13735c47c156463198b0905e7dad416e : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/af16dfb882904c9582aba01a28135429 : SUCCESS in 1h 26m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15edd9e0d1d340bf8c95b1b59f568f9e : SUCCESS in 1h 04m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/231fdfed7e38414598acd92f6a3b834c : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/386759e26fda49348b3a9d9925a035c7 : SUCCESS in 50m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4e03d6b156a4c299dee8cbf8a3626d3 : SUCCESS in 3m 15s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/d8b58ffefe06484f87285bfa59287283 : SUCCESS in 1h 15m 53s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/27371607f0144db2a60c43e4832b565f : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e7e15eb1643fbadcf4b04ac870eb2da8f30c18bc","revisions":{"8b925ee8536c287ee33e3a218dc630bbf2cfbfeb":{"kind":"REWORK","_number":1,"created":"2019-08-08 14:32:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/675353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/53/675353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/1"}}},"commit":{"parents":[{"commit":"6f5e0518dd055678b94df995ac7f841faa4359b4","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f5e0518dd055678b94df995ac7f841faa4359b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 14:32:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 14:32:20.000000000","tz":180},"subject":"Add mixed setup for debian based OS","message":"Add mixed setup for debian based OS\n\nSince ubuntu do not provide packages for tempest plugins, we are using\nmixed setup. We install tempest as a distro package, while installing\ntempest plugins from source, but without venv (for system python).\nThat guarantees that we have all required plugins and do tests.\n\nChange-Id: I198190d601cf8a922c3afaa69a3c0e42ab075735\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8b925ee8536c287ee33e3a218dc630bbf2cfbfeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8b925ee8536c287ee33e3a218dc630bbf2cfbfeb"}]},"branch":"refs/heads/master"},"818d5a0a97c0a4a5a6475a40bab05449a7f2076f":{"kind":"REWORK","_number":2,"created":"2019-08-08 15:53:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/675353/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/53/675353/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/2"}}},"commit":{"parents":[{"commit":"6f5e0518dd055678b94df995ac7f841faa4359b4","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f5e0518dd055678b94df995ac7f841faa4359b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 14:32:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 15:53:47.000000000","tz":180},"subject":"Add mixed setup for debian based OS","message":"Add mixed setup for debian based OS\n\nSince ubuntu do not provide packages for tempest plugins, we are using\nmixed setup. We install tempest as a distro package, while installing\ntempest plugins from source, but without venv (for system python).\nThat guarantees that we have all required plugins and do tests.\n\nChange-Id: I198190d601cf8a922c3afaa69a3c0e42ab075735\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/818d5a0a97c0a4a5a6475a40bab05449a7f2076f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/818d5a0a97c0a4a5a6475a40bab05449a7f2076f"}]},"branch":"refs/heads/master"},"c1f9ff4d3aa8951b95e5e23f9aa495d47b0a0b46":{"kind":"REWORK","_number":3,"created":"2019-08-08 19:54:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/675353/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/53/675353/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/3"}}},"commit":{"parents":[{"commit":"6f5e0518dd055678b94df995ac7f841faa4359b4","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f5e0518dd055678b94df995ac7f841faa4359b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 14:32:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 19:54:13.000000000","tz":180},"subject":"Add mixed setup for debian based OS","message":"Add mixed setup for debian based OS\n\nSince ubuntu do not provide packages for tempest plugins, we are using\nmixed setup. We install tempest as a distro package, while installing\ntempest plugins from source, but without venv (for system python).\nThat guarantees that we have all required plugins and do tests.\n\nChange-Id: I198190d601cf8a922c3afaa69a3c0e42ab075735\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/c1f9ff4d3aa8951b95e5e23f9aa495d47b0a0b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/c1f9ff4d3aa8951b95e5e23f9aa495d47b0a0b46"}]},"branch":"refs/heads/master"},"98b2e7a2ac923ad69ca0a904c5546ba7c697c860":{"kind":"REWORK","_number":4,"created":"2019-08-08 19:58:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/675353/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/53/675353/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/4"}}},"commit":{"parents":[{"commit":"6f5e0518dd055678b94df995ac7f841faa4359b4","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f5e0518dd055678b94df995ac7f841faa4359b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 14:32:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 19:57:37.000000000","tz":180},"subject":"Add mixed setup for debian based OS","message":"Add mixed setup for debian based OS\n\nSince ubuntu do not provide packages for tempest plugins, we are using\nmixed setup. We install tempest as a distro package, while installing\ntempest plugins from source, but without venv (for system python).\nThat guarantees that we have all required plugins and do tests.\n\nChange-Id: I198190d601cf8a922c3afaa69a3c0e42ab075735\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/98b2e7a2ac923ad69ca0a904c5546ba7c697c860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/98b2e7a2ac923ad69ca0a904c5546ba7c697c860"}]},"branch":"refs/heads/master"},"e7e15eb1643fbadcf4b04ac870eb2da8f30c18bc":{"kind":"REWORK","_number":5,"created":"2019-08-12 09:46:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/675353/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/53/675353/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/53/675353/5"}}},"commit":{"parents":[{"commit":"6f5e0518dd055678b94df995ac7f841faa4359b4","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f5e0518dd055678b94df995ac7f841faa4359b4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 14:32:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-12 09:44:29.000000000","tz":180},"subject":"Add mixed setup for debian based OS","message":"Add mixed setup for debian based OS\n\nSince debian/ubuntu do not provide packages for tempest plugins, we are using\nmixed setup. We install tempest as a distro package, while installing\ntempest plugins from source, but without venv (for system python).\nThat guarantees that we have all required plugins and do tests.\nTo accomplish that variable tempest_plugin_install_source was\nintroduced.\n\nChange-Id: I198190d601cf8a922c3afaa69a3c0e42ab075735\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e7e15eb1643fbadcf4b04ac870eb2da8f30c18bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e7e15eb1643fbadcf4b04ac870eb2da8f30c18bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
