)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~708844","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I9ea47452c430c6bf5ec0ac6332cecd5d3abc7792","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"tempest-worker-path","hashtags":[],"change_id":"I9ea47452c430c6bf5ec0ac6332cecd5d3abc7792","subject":"Add support to tempest --worker-file option","status":"MERGED","created":"2020-02-20 10:25:08.000000000","updated":"2020-05-04 23:23:03.000000000","submitted":"2020-05-04 23:21:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708844-1588634495294-57c03af8","meta_rev_id":"2697665474be98b8af6e540f596c01a56b01a7c4","_number":708844,"virtual_id_number":708844,"owner":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 23:21:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-04 16:51:20.000000000","_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-05-04 13:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":2,"date":"2020-05-04 20:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":1,"date":"2020-05-04 20:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 10:42:33.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-04 13:58:38.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-04 20:58:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-04 23:21:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73d8c9618030c70edf08fe93358058b3f59425e5","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-02-20 10:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79fd69599b06efb1d9d5b59ea8ef688b12d421d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 10:42:32.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"978802b6b66f73a465d898642ecf42df282e7c07","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 10:42:33.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"fe7e8b3bee21d09dc79fe92e18d885b5021845fd","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-05-04 09:27:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cc1648b112d360ef2eedb4da15ce4c1a5a665a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:29:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2abf6b1db7344ec99adf1304bacd7310 : FAILURE in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9ec3383575f441d8a1fb2b2f214d8d00 : SUCCESS in 1h 06m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/911884af33ad45acb6d57731af475aa7 : FAILURE in 18m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4a860ab6b28845f0b0abff9dba2d78d8 : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c91afd4411e1471b93cf9c20b856a330 : SUCCESS in 1h 04m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/48da2293fd144811814a7c90956cdb62 : FAILURE in 29m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b050bd82c3c49be86b4dab970ec4aa8 : FAILURE in 7m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7071cf38f50942ac90d0c84c382cbbb6 : SUCCESS in 1h 08m 42s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0aa42d384d964b8f8371bd248f789cdf : FAILURE in 33m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23f83256afb34d69b203607ab4cd5bb1 : SUCCESS in 53m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b562fcb25f1f4f7d8fe793d3189c980b : SUCCESS in 5m 27s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/022208986fd34fd08d9e70b60dc0144d : SUCCESS in 1h 07m 09s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3e3c9595b830434cb1c938f68a9ed03b : SUCCESS in 5m 35s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6df7be8135ef42afa21102774580fa9f : SUCCESS in 1h 56m 50s","accounts_in_message":[],"_revision_number":2},{"id":"0f050bf2e2dc367d75399bdf72ff2328b1e37181","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-05-04 11:50:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e1d7ec985fe32ddd79afb124b10795d5f58422ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:46:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4b01460b89a0471e966870518476f112 : SUCCESS in 6m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bf4450b1fb16424db50fa062ace0307d : SUCCESS in 1h 14m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6fabd56891994b968f017580def874a6 : SUCCESS in 1h 30m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2b711c9f0dea4efcb9ddc0e807222902 : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cfb69b88d9b42258cb2bb0530b34a0a : SUCCESS in 1h 15m 51s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fd031cf0f310458e9368a9198c4599f3 : FAILURE in 22m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22da3603dbed4130bc25560e753c12b8 : SUCCESS in 1h 50m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/59890ed530f94770ae9aef659894e49f : SUCCESS in 1h 03m 18s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a955f153e7b94b33839b0ebd486dd8b6 : FAILURE in 38m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb2cc008462c4c2a92f4671364c4008b : SUCCESS in 1h 05m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4682b8f9a25c48ffbb7ddd0f7fee11d6 : SUCCESS in 5m 32s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/b42edab06d5a4a0c99f6faef243a9e9b : SUCCESS in 1h 15m 24s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/46227572b2e64555a26a3f7aba6aa608 : SUCCESS in 4m 47s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e2bab9a398e94d51876e2e0e57b84a73 : SUCCESS in 1h 45m 20s","accounts_in_message":[],"_revision_number":2},{"id":"bdcf79fe9524e7075abae6582d84edbcc5e4357b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-04 13:50:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nJust one tiny nit","accounts_in_message":[],"_revision_number":2},{"id":"22a5b1ecfb64aeef54be5f574d125c8b9eec0cf4","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-05-04 13:55:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3965dcd8f4fcb8bcc380c501d88cf8f2cd7f2edd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-04 13:58:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"175558ba22d0aee698965e1c13d91278c08860a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:41:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a46a8512f7be4988b1ab3d7e3265e968 : SUCCESS in 11m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c9c05ca8e7a34f538799212a2378f92d : FAILURE in 1h 49m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f80747e9e01c44d7b5ea374e893244ea : SUCCESS in 1h 52m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/da58f7398c784ec1ac8642e9bef4769f : FAILURE in 23m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc8e19ddbcd14d7bb8a770c0d60097ab : SUCCESS in 2h 07m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/091817a3e32a4ed0b25b68790708db67 : FAILURE in 29m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d6e062df9c849d1b2fd86a81d94b023 : SUCCESS in 2h 41m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/91942e6779db4c40b8769895c2a89a38 : SUCCESS in 2h 00m 43s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f1b83d8968fc4c53b53d8ee415d06e26 : FAILURE in 29m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f09ddca09d0404f9deb27daf1d0a281 : SUCCESS in 1h 58m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c92b9bfc6f466fac5ef1302a6b3893 : SUCCESS in 14m 30s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/43bc4aefcef444b989eb26a5be7c951a : SUCCESS in 2h 01m 51s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2f40026a4118489e9d37f9f4b7be98ae : SUCCESS in 9m 28s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9348d1e0d47a42518b04c4dba3be4457 : SUCCESS in 1h 53m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0b0c5fd6c31ff4d21dcdada4723cb7bb16c2607a","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-05-04 16:51:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1ce4fbf24aa55f41a80912e8afff10b0b1c55259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:53:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b9195a29eb884a62802dbff769375d0b : SUCCESS in 6m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cb8fd93187ff451dbd630399c5a164f4 : SUCCESS in 1h 06m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/74da4f0e5c69489e846665c0ce714bde : SUCCESS in 1h 36m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4f18f896223548f387056d4a1b3a8b09 : SUCCESS in 1h 35m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7ef0f35468543cea548bb9cfea89fba : SUCCESS in 1h 13m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a6d29e19eb00466e813ba9b3882e2cb7 : FAILURE in 24m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc28048ea04449bbac73a4b3bc14241c : SUCCESS in 1h 57m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ab98467190f2438797650dcac36cc3c3 : SUCCESS in 1h 09m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/56f44e8f9fe04f5ea876f899a4f13bea : FAILURE in 36m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce451c919fad4df281e0515c9a585dfe : SUCCESS in 55m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cebcbbc4ee94da5a72b76674461a775 : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/620a4a983f3641aa8311eafcd537d4f6 : SUCCESS in 1h 51m 52s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/977a65d5524a40c7a72ef9ce9b3ba223 : SUCCESS in 6m 35s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/fac1d4fa352144d1ad4efe5f12e13e02 : SUCCESS in 1h 41m 13s","accounts_in_message":[],"_revision_number":3},{"id":"1959296eec7d603dc9296c87e7bfe02cd1c98591","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-04 20:58:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"707b56f33faf76640d41845f793503d672cb36a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 20:58:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2ec5038ab1b78e25baf7f02bf5dc36891a91e1f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 23:21:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0383aa4a39884387a592ceea531ab15b : SUCCESS in 12m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0157c2c8d7ad46359d01da1c59713610 : SUCCESS in 1h 09m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/95de025bfebc46bb90fae78a6f4fd167 : SUCCESS in 1h 44m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5d9dc170cd234934b205fb311b167871 : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4f7f9dbfaf64dba825a91f2455a5633 : SUCCESS in 1h 13m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6eb8965182ee4da8b77b441fae755f2d : SUCCESS in 1h 14m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8e81644a1384cc5af88425ac46aa9de : SUCCESS in 1h 21m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4508c3330c104ec1a0cf30964d63234e : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/c5602d4b427348a5b8885a70bc472cdc : SUCCESS in 1h 22m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8183bdacd7f24c6cacb0978287ab8fab : SUCCESS in 2h 21m 51s","accounts_in_message":[],"_revision_number":3},{"id":"5f6a06a7034fe238443a253358f487cfc79bd9cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 23:21:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6112d5f27ee518b4b8049326cccac68cb86f7423","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 23:23:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e178ab1144db4d57ab09f9f88330db4c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e6476f5c1c667bde34eda8a29623df3dd7dbd99","revisions":{"eae607bc9511fb0a51dca0c035ec6be2bb0221cf":{"kind":"REWORK","_number":1,"created":"2020-02-20 10:25:08.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/44/708844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/44/708844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/1"}}},"commit":{"parents":[{"commit":"81507e82607882b0ddf553c85647f4151c813ea8","subject":"Revert \"Mark ubuntu distro install as non voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/81507e82607882b0ddf553c85647f4151c813ea8"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2020-02-20 10:25:00.000000000","tz":60},"committer":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2020-02-20 10:25:00.000000000","tz":60},"subject":"DNM: Add support to tempest --worker-file option","message":"DNM: Add support to tempest --worker-file option\n\nThis patch add support to tempest run --worker-file option.\nThis add the possibility to manually schedule how tempest tests will be\nexecuted in each worker\n\nDepends-On: https://review.opendev.org/#/c/708833/\nChange-Id: I9ea47452c430c6bf5ec0ac6332cecd5d3abc7792\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/eae607bc9511fb0a51dca0c035ec6be2bb0221cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/eae607bc9511fb0a51dca0c035ec6be2bb0221cf"}]},"branch":"refs/heads/master"},"8ed041c175bc3ab7ed7e3b58a8cf96e1b960f653":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-04 09:27:37.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/44/708844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/44/708844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/2"}}},"commit":{"parents":[{"commit":"81507e82607882b0ddf553c85647f4151c813ea8","subject":"Revert \"Mark ubuntu distro install as non voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/81507e82607882b0ddf553c85647f4151c813ea8"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2020-02-20 10:25:00.000000000","tz":60},"committer":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2020-05-04 09:27:34.000000000","tz":120},"subject":"Add support to tempest --worker-file option","message":"Add support to tempest --worker-file option\n\nThis patch add support to tempest run --worker-file option.\nThis add the possibility to manually schedule how tempest tests will be\nexecuted in each worker\n\nChange-Id: I9ea47452c430c6bf5ec0ac6332cecd5d3abc7792\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8ed041c175bc3ab7ed7e3b58a8cf96e1b960f653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8ed041c175bc3ab7ed7e3b58a8cf96e1b960f653"}]},"branch":"refs/heads/master"},"0e6476f5c1c667bde34eda8a29623df3dd7dbd99":{"kind":"REWORK","_number":3,"created":"2020-05-04 13:55:48.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/44/708844/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/44/708844/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/44/708844/3"}}},"commit":{"parents":[{"commit":"81507e82607882b0ddf553c85647f4151c813ea8","subject":"Revert \"Mark ubuntu distro install as non voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/81507e82607882b0ddf553c85647f4151c813ea8"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2020-02-20 10:25:00.000000000","tz":60},"committer":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2020-05-04 13:54:18.000000000","tz":120},"subject":"Add support to tempest --worker-file option","message":"Add support to tempest --worker-file option\n\nThis patch add support to tempest run --worker-file option.\nThis add the possibility to manually schedule how tempest tests will be\nexecuted in each worker\n\nChange-Id: I9ea47452c430c6bf5ec0ac6332cecd5d3abc7792\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0e6476f5c1c667bde34eda8a29623df3dd7dbd99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0e6476f5c1c667bde34eda8a29623df3dd7dbd99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
