)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~709977","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I69587a455cb875e5e1f1a509a87082ceeab52d91","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"fix-stackviz","hashtags":[],"change_id":"I69587a455cb875e5e1f1a509a87082ceeab52d91","subject":"Use current stackviz tarball","status":"MERGED","created":"2020-02-26 08:27:58.000000000","updated":"2020-03-23 19:54:58.000000000","submitted":"2020-03-23 19:53:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709977-1584993207670-dfe5b5ea","meta_rev_id":"63afc11606380fce53d83d68436a38e4db57e22b","_number":709977,"virtual_id_number":709977,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"date":"2020-03-23 15:29:23.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 19:53:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-26 09:41:02.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2020-03-23 08:30:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-03-23 13:53:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-26 11:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-03-23 13:53:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 09:41:02.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-26 11:14:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-23 08:30:43.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2020-03-23 13:53:06.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-03-23 19:53:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d8b51703f30e0a613fe60fb759605230ed3312e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-26 08:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"828f28f84388f20ce1aa059c9c9702e6f58e5114","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-26 09:41:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-rhel-8-standalone-rdo https://review.rdoproject.org/zuul/build/e0c40f520fcb4a6e85fe112ee8018f9c : SUCCESS in 1h 12m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f2c5f0dfa964da9fafb3aee1491efcaf9a8ffea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:25:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d3533c88b0944b90b53be990bbf03871 : SUCCESS in 8m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/93bfa97c22124fabb2c27cad0f467817 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1196467ad74346b6a004e64c66aa63b6 : SUCCESS in 1h 30m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9d626a183b254eec8d1adefdbf3f8eeb : SUCCESS in 1h 24m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfe00522d60044f498bb1e99d08dbbf6 : SUCCESS in 1h 15m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/56a608b1eebe448f8ced7f308958c4f5 : FAILURE in 1h 18m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35fd2237472248c7bf05ed8b937082e0 : FAILURE in 59m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/59b9808f858545cca9e18f78b7440fa1 : SUCCESS in 1h 08m 16s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/82c06ee708fb47de95c7751fde9119d8 : FAILURE in 32m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/307d2679d64e419393f4f50c8ca3238a : SUCCESS in 1h 03m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a13f8ff0be3c4011a6e55dc8b684c944 : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/f27ac90cb97e49a386ce06aca4ead27e : SUCCESS in 1h 14m 41s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6bd39ce2792d4d77b7e4bf2671e4866f : SUCCESS in 1h 53m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2cb789a81777d4cb0d8d53a42707a3c02b66e8ec","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-26 11:14:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01608fd7cc2fd349f72b34714737c8ae02c81a9d","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-03-23 08:30:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"90ef64ec8ba909325ad16a06be2fa2a634bbf3df","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-23 13:53:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06e443770d9774b8b2e32fbde3a7cc5e6a3ac8db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 13:53:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2fee1cc277462b88586ec77ba5708cff7837e04a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 15:24:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0d932b3453e04bbf8c8b6e921c4f4d04 : SUCCESS in 7m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/833da178f5c14be98f1d5987f33c50bd : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/74433bd96a90420e94497830971a1811 : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/62b24fc7b3974addb0274e5a570f9a56 : SUCCESS in 1h 24m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/664b16d500344b728b314671857796fb : FAILURE in 37m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/da929c0546a74ac1880ed81dd27d8eac : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/720903ecb1e14377bfbe3c438bd3767f : FAILURE in 39m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e89bdc3c02847e9b4ab1d37774adf36 : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/77af886769d543d4b7e17a6608df37c6 : SUCCESS in 1h 07m 41s","accounts_in_message":[],"_revision_number":1},{"id":"169b7ae5c073c6c77a60ee052cadffd446ed718a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-23 15:29:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"638cf774f992c2b63917ed43f398644c3c450412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:45:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fa64f2f87ff146e596c2eeecbc9985a5 : SUCCESS in 9m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f46d670b5a1e4b6bb8d127147eab7d35 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5ed6fcf9a6824d67938dc2ffb631d8de : SUCCESS in 1h 47m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/50d51e3a6cb54f8083c041ac8f3f42c8 : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bcf020cfe13401488bfc08660241bda : SUCCESS in 1h 33m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f89e0646240c4a33ba9e6a0587192a35 : FAILURE in 8m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fe8f139df164ac4b149c6dfc4ccbc7a : SUCCESS in 2h 09m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6b659f07422545e5afb824a99bd0528c : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d5f198e9537c4218b3534d1396c56542 : FAILURE in 32m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de870cd2b2aa42148ccbc2a6ed2ec9c8 : SUCCESS in 1h 20m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26ecdd0bf15444ba20c12a72d11a253 : SUCCESS in 8m 05s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/7cbbba792cde4b23a358aeade5a9da48 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cb8cfffbbf92aadc44f506068a729ae909394a44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:46:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef783cf5956ee65c2c8b81517be2811950baa1d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:53:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cab18e4a33914590a28c0f4712972e52 : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b89a087654284fbd8ac2a63f5ff7a4f1 : SUCCESS in 56m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fba030d23b1c422c978a34d3559cb007 : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bed0b2d2edad46b18210c5d3558c00ee : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a9e0edfc200461abed16b5ddddaeb04 : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5f6108e902394e8d9149352c9ecd7e4a : SUCCESS in 1h 06m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/902f32bb25e345fea2127466d5f36ee1 : SUCCESS in 1h 14m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62a3a21c2494591b739e15bb2097c51 : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/8cbe08a8459f4eb3838f709e7e17b4c2 : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3a7e7061a389b0283cf79e129ce3fb2f951a061c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:53:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"699ded795ad2e029029bb35c820f9a04caf5867d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:54:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb3c0f908964a6f9cf0223987aa3c4d : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35f80371c2fc33e0e6c3f3968d6fbd017d8a72e7","revisions":{"35f80371c2fc33e0e6c3f3968d6fbd017d8a72e7":{"kind":"REWORK","_number":1,"created":"2020-02-26 08:27:58.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/77/709977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/77/709977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/77/709977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/77/709977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/77/709977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/77/709977/1"}}},"commit":{"parents":[{"commit":"7f131f99606c829421e3d820bf5976f5a9d934ee","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/7f131f99606c829421e3d820bf5976f5a9d934ee"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-26 08:26:33.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-26 08:26:33.000000000","tz":60},"subject":"Use current stackviz tarball","message":"Use current stackviz tarball\n\nhttps://tarballs.openstack.org/package-stackviz-element/ has not been\nupdated since 2017, use the current location for the stackviz tarball.\n\nChange-Id: I69587a455cb875e5e1f1a509a87082ceeab52d91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/35f80371c2fc33e0e6c3f3968d6fbd017d8a72e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/35f80371c2fc33e0e6c3f3968d6fbd017d8a72e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
